)]}'
{"id":"openstack%2Ftrove~803126","triplet_id":"openstack%2Ftrove~master~I05a07402388d214dc95071be709191f63af49411","project":"openstack/trove","branch":"master","topic":"reset-master-for-replica","hashtags":[],"change_id":"I05a07402388d214dc95071be709191f63af49411","subject":"Run reset master when setting up mysql replicas","status":"MERGED","created":"2021-08-01 23:17:49.000000000","updated":"2021-08-02 12:11:10.000000000","submitted":"2021-08-02 12:09:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803126-reset-master-for-replica","meta_rev_id":"ce4d4193c3eaa9d313e0dab0d45e602eaff498d9","_number":803126,"virtual_id_number":803126,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 12:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-02 12:11:10.000000000","post_submit":true,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-02 09:43:17.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 01:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b334147ce01980755f9481a21e139504993d2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-01 23:17:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff042093dbbc39468f178cace5f050550a618b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 01:06:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f87873280c4f698865c6f4c25da680 : FAILURE in 7m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab38eba50824fc6b9b50dc837847e37 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c0722bbbd694184862ade962579bdc3 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84020b06afc84561b98e757b4bbb6466 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84bfc8af6b8c4114a0ea06782b0e4e43 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f828422995e424c8b54389d10ffd693 : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c14d6dc356f34bfa82a8aaaf9678b02c : SUCCESS in 6m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cf5df85645cb4e28ab66249a8e7d86ee : FAILURE in 1h 17m 07s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c606be0938064664bd8a883b1358b87a : SUCCESS in 1h 28m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f336882050246d4860b521ac96f0027 : SUCCESS in 1h 43m 45s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6dab3b0c87e94c76908115fc8465f5cc : FAILURE in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3a45540cc0d6519ae53d444bdec62e872d66f34","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-02 02:28:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cae6b61f572687caca62d16c0c148de2fca262f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-02 09:43:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"79b58ef070ba30516c37e0372b5f347959bfa380","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-02 09:43:17.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e199c0930f4157c6371c6f7d27748776bbdffe9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:35:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4a0ab029a844b54af6b53d93e077717 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5874bab3f57b4bacae7275b359fefaec : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc7fcd5f9cbe4dffb0663eab62997a8e : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae68cd9f3db40b199e2947a4d1350f3 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48a72bf213764ba582843c007c44f33c : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7811f7dd0d4a338f0ab1fb29227b67 : SUCCESS in 6m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/016d0adf76874775b954fead642b36b1 : SUCCESS in 6m 31s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d708eec4759b4860b77f8504e57aa102 : FAILURE in 1h 14m 36s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/777479b1fab74b598f499f089913145a : SUCCESS in 59m 22s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aff2f8d4fc64620ba0283cda0a0baf1 : SUCCESS in 1h 46m 59s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7c50e144fe044e40a41d078da53aef6b : FAILURE in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3072bbd65c4b746f691f22767bffd58ace9e6ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:36:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"314fd0618824743bca067052205c18ca894ed619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:09:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156b8df4fcb045a0bb86fe4ff1ffae24 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b74f8f11a4154590b5aff3fbc6de5c29 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c348103873f4c678e8998b772f39424 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6a0b82c83f4fa3994fa415c88e3521 : SUCCESS in 6m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d03670c534094993a37d6f91ad3f8396 : FAILURE in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0251750d5740a8aede474ac688e4c76a67f2f14b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:09:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e3bbd256428206477548e2baac29922b68a4bfe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:11:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e254488cdb146db9aa38c68441bb542 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a12b2be7060f6a98d5ff52a0823b40f012a31449","revisions":{"29246c76fae0c248763aec9f0f9e6c15336bc766":{"kind":"REWORK","_number":1,"created":"2021-08-01 23:17:49.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/26/803126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/803126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/803126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/803126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/803126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/803126/1"}}},"commit":{"parents":[{"commit":"5bba60704547202b1e3d535f732aa345ffe92c8c","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5bba60704547202b1e3d535f732aa345ffe92c8c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-30 04:20:08.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-01 23:17:22.000000000","tz":720},"subject":"Run reset master when setting up mysql replicas","message":"Run reset master when setting up mysql replicas\n\n* Only run \u0027FLUSH PRIVILEGES;\u0027 when needed\n* Fix SqlClient contextmanager initialization\n\nStory: 2009091\nTask: 42921\n\nChange-Id: I05a07402388d214dc95071be709191f63af49411\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29246c76fae0c248763aec9f0f9e6c15336bc766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29246c76fae0c248763aec9f0f9e6c15336bc766"}]},"branch":"refs/heads/master"},"a12b2be7060f6a98d5ff52a0823b40f012a31449":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-02 09:43:07.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/26/803126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/803126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/803126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/803126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/803126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/803126/2"}}},"commit":{"parents":[{"commit":"036948c516672e55c4a2c932c74bcdec80319108","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/036948c516672e55c4a2c932c74bcdec80319108"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-30 04:20:08.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 09:43:07.000000000","tz":0},"subject":"Run reset master when setting up mysql replicas","message":"Run reset master when setting up mysql replicas\n\n* Only run \u0027FLUSH PRIVILEGES;\u0027 when needed\n* Fix SqlClient contextmanager initialization\n\nStory: 2009091\nTask: 42921\n\nChange-Id: I05a07402388d214dc95071be709191f63af49411\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a12b2be7060f6a98d5ff52a0823b40f012a31449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a12b2be7060f6a98d5ff52a0823b40f012a31449"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
