)]}'
{"id":"openstack%2Ftrove~803132","triplet_id":"openstack%2Ftrove~master~Ia1e9112ae69e04f8c3e9e9d1b4a0189c743d7448","project":"openstack/trove","branch":"master","topic":"show-networkid","hashtags":[],"change_id":"Ia1e9112ae69e04f8c3e9e9d1b4a0189c743d7448","subject":"Show user network ID for getting instance","status":"MERGED","created":"2021-08-02 02:26:07.000000000","updated":"2021-08-02 11:23:56.000000000","submitted":"2021-08-02 11:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803132-show-networkid","meta_rev_id":"da24720bed7318501855cb7cd32779618267379e","_number":803132,"virtual_id_number":803132,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 11:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-08-02 10:19:18.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-08-02 10:19:18.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 03:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bafcd63a48bee067440ab5f2f3e372eb9f842758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-02 02:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4baf54e2bac91df9c6af3285011a6680cf80e79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 03:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9dd3de52593c444a9728a56c7f0bfcba : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/045597890ed9405483a622d2e359c6b2 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf2e43300e3c4768930dc28470abcfbf : FAILURE in 7m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b55b1d6570c4cd4a7351c0e66ad4e44 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14541f5be3864789b8138aa5c4db8b5e : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08d368fc1cc440a99ff24ad3ac97e709 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80ce86f31f69438cb05972ba001f4fe0 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ae8e88d31b495fb8d4481230eaa5f5 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd33127dbed147e3be4ad0e47e849bac : SUCCESS in 6m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a1d75f942a84bdebfde029c4a663255 : SUCCESS in 6m 56s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/424ad088ee084bd899dc3855dd52eca9 : FAILURE in 37m 49s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d6360df31ba5454ba15cb8f05ff46261 : SUCCESS in 59m 24s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b7917618dbb4651b9e4b650658c4ec3 : FAILURE in 1h 22m 25s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/06a90548194e4a16a7061dbb2e1b363f : FAILURE in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cef36cb7958d0758fcfa735c3664c5e2df906748","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-02 10:19:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a4fab8d6a25aeb89743b8747b266ea39e68c6029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:19:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2566c4e642940e90251aad5e5d343d925c4db90e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:22:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ae8615a1d4d45c7a6949c31eabde430 : SUCCESS in 6m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/40059ca138934af7af5558e7e76a3eeb : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee9ed4cbed546dd9427deb76e946f1a : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e96b6b8e8b14a93a82e506909c328e5 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a0677b8e3164c589baa8ce0e43da2d0 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ab19c5a05d4fc8a1d8aa51d68c13aa : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9922bac7e66845a5909a38e222cd229c : SUCCESS in 7m 44s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5c22672513804e008bf00e33b04bf2d6 : SUCCESS in 1h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5405c22ef2df98d07cb5a82151fcc0e88e207d55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:22:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da24720bed7318501855cb7cd32779618267379e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:23:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4054dbe7700b45a0a3f742fac70e7266 : SUCCESS in 1m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1dd4b7f7ab43d1ac5f15d74d7086a2 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95a9cb718e34716ba340294a6d43ba6 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5590ecdce036f7a2b11ff440dfd4eb637180abcc","revisions":{"5590ecdce036f7a2b11ff440dfd4eb637180abcc":{"kind":"REWORK","_number":1,"created":"2021-08-02 02:26:07.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/32/803132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/803132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/803132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/803132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/803132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/803132/1"}}},"commit":{"parents":[{"commit":"036948c516672e55c4a2c932c74bcdec80319108","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/036948c516672e55c4a2c932c74bcdec80319108"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 01:15:56.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 02:25:54.000000000","tz":720},"subject":"Show user network ID for getting instance","message":"Show user network ID for getting instance\n\nChange-Id: Ia1e9112ae69e04f8c3e9e9d1b4a0189c743d7448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5590ecdce036f7a2b11ff440dfd4eb637180abcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5590ecdce036f7a2b11ff440dfd4eb637180abcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
