)]}'
{"id":"openstack%2Ftrove~803192","triplet_id":"openstack%2Ftrove~master~Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82","project":"openstack/trove","branch":"master","topic":"devstack-ovn","hashtags":[],"change_id":"Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82","subject":"Use OVN as the default Neutron backend in DevStack","status":"ABANDONED","created":"2021-08-02 11:14:24.000000000","updated":"2026-09-18 10:48:39.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5d3bfc92fba55230396534bfe80eb0f27cc49d7e","_number":803192,"virtual_id_number":803192,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-10 03:15:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 11:36:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-08-02 13:20:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51e6548db0f00fa0de1693e6a90210df29a02586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-02 11:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac3c0306b70ce7a2c7239cf1cdaa3ac587399403","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-02 11:36:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"345a4a55699c58d24b450c52adaba518dfb5ad44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 13:20:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/890f69ecc8684c8f9b7e91b75cbf83f4 : FAILURE in 6m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527c9262ac9e48519f46369a59011af0 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9db4351f50434384a7e75ac761292c69 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/867698be407f419aa4bdb6166f929f04 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec0accab410e4411bd8eba603778767b : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/655fcf4a83ab4087b4966d1b0bc92af5 : SUCCESS in 9m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a61f53ea66a47cba0adb2f304d5e582 : SUCCESS in 6m 30s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ef4cff514b1046b0a42135a4dd591677 : FAILURE in 1h 27m 24s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a5698c9e13d7424ab51fb2dc9e0a30ed : FAILURE in 58m 47s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e3dbee6f2c04a1590ec08161079acef : TIMED_OUT in 2h 04m 37s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/26f003ad9090442a93d0df2c9fd1e50f : FAILURE in 36m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d8af87c4b917f9d295b084fe34ddc89ce9619ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-04 00:55:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"139fa460b88ef2f5ed9ba62134cf3bbcb7d1d583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 03:01:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/491dba413358412a8666fc5fb9bf6dd4 : FAILURE in 8m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de01ffe572344180831e4ef8c81f7585 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ed10d2714eb4bafb956191e4b8eb435 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59d90ebc1ace4ddfa8b2cdd2008e95d9 : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdec8d0618fc482e865227de607b3a6d : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b835eaf7a3db489185aa8728932709d3 : SUCCESS in 6m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/907af177fb404b91925d758258720430 : SUCCESS in 7m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/beaef8fc52bb4b9bbe3e26a887ae9414 : FAILURE in 46m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ef85eda8fc9a462fb1e44cc67495f4ae : FAILURE in 45m 38s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e87c8da8a1c4c26b884094c3c148da8 : TIMED_OUT in 2h 03m 24s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/02825bdfbf844599b0ef3a73d6d076f3 : FAILURE in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5233f3c66c0adc30c9645fd616ba96d057d47f18","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-04 09:08:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2dafaec276cd041ac05b6bd15980fe7b6089a70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 10:45:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7516813b7b4f4101b4634c3025e34b56 : FAILURE in 11m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a1b8aed0f24c9e8a795d8d85590e73 : SUCCESS in 9m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78d4200e8f594145882ee57a1472b336 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa5129cf10444bee90422a0d5ed5a854 : SUCCESS in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acef1be4062f4a36a6986ddf91945554 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91985cbe1ebe4c6eadbec0ffa6616faa : SUCCESS in 6m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d403c75d30704bf6afab499649bf345d : SUCCESS in 6m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/882b868c7fc042c7a3824cd05213f92b : FAILURE in 43m 02s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/faf43e05b01a4587ad6e4e1386c43a61 : FAILURE in 47m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1277af3f88f043a38bc338c2e394e04d : FAILURE in 1h 30m 25s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/18a00068e4f246cbbff232304fe23613 : FAILURE in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6eaea78706060049a60386382ab17bdce3e0df5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-05 23:11:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"35f776df1e59da701eaefea513b94a1f9e55665f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 00:34:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96e235eafbb6442688fd2de8ab0c6792 : FAILURE in 7m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e791ace09b1a47cdbfc3dad365bb90ad : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42d9f0b8277048f4bca34100761f64ac : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8ed7e7db47d49ce9704591ef9046048 : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/425b05dd03a04ef78b0a7b8805be51a7 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9b84e9932447c5a9fb44edb173aac6 : SUCCESS in 6m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38364cbd7b384acb8ced186e4f8f4003 : SUCCESS in 6m 25s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/984b0de7a98c4392a0998efd8deb9411 : FAILURE in 55m 30s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a5fd44ffbf104a578a01d3a167b95a70 : FAILURE in 55m 06s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19dd1a17cdcc4ae5a3ce3c1f1831686a : FAILURE in 1h 13m 23s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/56ceba70668647c3acb39c50d09e2ea7 : FAILURE in 41m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"928758b808143274652164c5060c763838b61509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-16 00:47:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"97d173292a01bf2eb9814a6233fbf11575d00057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 02:07:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/248d2f600773499496a9ae816982d4e2 : FAILURE in 7m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4435621aa221438ca3b1356be209d095 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cfb6a3159054cf985962629ce47177f : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8208f5a0906a48ebb948b5499b66b7ba : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d0ddab2e8ee4a668166b5e3e9c12d0f : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a778b18759524c449aba4c8db0ec3075 : SUCCESS in 6m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8432097c4c744dbe859e05255e75670a : SUCCESS in 6m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/55c6d45757874fd2bded4ea1c0f1b5ee : FAILURE in 42m 35s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ec068ef440f74b29b227d28954004e37 : FAILURE in 48m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4580aa635e4f50bbeab8843d52ef80 : FAILURE in 1h 14m 00s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d03b94470cf442299ec600fe8b977ef2 : FAILURE in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8b99f9302f3ceb6a8e3f09122e8da60d4e23a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-10-09 19:19:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a80882b19ed2b0613f1c5818bead6ce273280dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 20:21:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b32b8cef9b3f44399c4ad2e85345ebfb : FAILURE in 7m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a007f79a87e40939015fa506367d5a4 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/859a8ab3ceeb431ba97a95b5f85d6882 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064d05e6043546dcbe210cecd686f96b : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60b097429c404d638ee99592e6ac021e : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd74be825c2b41e881b245fc414c8956 : SUCCESS in 6m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5585823e4b844f04b9b6786185eb83fd : SUCCESS in 6m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/98964d27b37f45738db8f9908f08d72a : FAILURE in 47m 22s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/57b62fcef634454185d2b7b189d704b4 : FAILURE in 48m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/844b3226e0ab43679933f54fcb6633ec : FAILURE in 1h 01m 09s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/623ef498ad5d4220bb5283b7ee6afd2c : FAILURE in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"18b8a19552937896a47eb9629e802b12ef5d380f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-10-10 01:22:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fbf43678f95f7adcab252834668c723134fab472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 03:15:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a91d7e1602d49d688cd4f7d3efb8fb7 : SUCCESS in 14m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26e59fdaf2464d1a9f5a6a071ea63c3d : FAILURE in 9m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bea2954c2ce456d93912466c5f8ca19 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3ca18e6ac37468d974d11c1669bdfe1 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cd4049b6db44a2a9f426f02ef21ef17 : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed194d10e7441518c9149800879f056 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9752326e10de4c60b14d08a5d9e63f82 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/708f6bec6b3a416581756bd11bf57af5 : SUCCESS in 8m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d88978de337438dbad37f8c7fbe3a80 : SUCCESS in 5m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8f4741bc14bf48fd95da428b9aaeb7a1 : FAILURE in 56m 03s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/03c5af1b56c64518b0bcd52e7745530b : FAILURE in 49m 08s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c66fd8b693f425cb512211e90dc7814 : SUCCESS in 1h 44m 38s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/85d06b20e730411ba94c608351b9c120 : FAILURE in 30m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"998e15730a63b9efe5add1254074f2f8ab360c80","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-10-11 21:47:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ffd43d51293cce2c144cc802befc4f94477c5536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 22:45:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59f31baf7d6f4926a565bd1b93b494df : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b7c0a95ae634f19a7d37ec9f7b95d6f : FAILURE in 7m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb578658d6e46e990ea1d864f81749b : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c4c2b0260fa4473bec4fe21a01e9c9c : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff7b03e1fc7d4fd6abd69858ffc45e6f : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33f471ba5b89482282e20174281f88a5 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086bfdb1723347cb8d9c10cace24817c : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7390afa7f15247a681a4ccfe055c9f20 : SUCCESS in 5m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc01638a6c96444da52f4772ef4c5052 : SUCCESS in 7m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9b4878e3b72c468382e1c8313700ac68 : FAILURE in 55m 09s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d6bdc9b46bd641f488cfbbd171d40a04 : FAILURE in 44m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4014265f0024681b39fdb0bf8c770bc : FAILURE in 32m 13s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/33c76dc7c1a54a1c8f2c85845d8c2804 : FAILURE in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5d3bfc92fba55230396534bfe80eb0f27cc49d7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 10:48:39.000000000","message":"Abandoned\n\nAlready implemented. Cleanup.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7997fe91a9161a58816f65e2663615cd230a7074","revisions":{"2b83494c8d32506b1db88aebaa72c2330d08b383":{"kind":"REWORK","_number":1,"created":"2021-08-02 11:14:24.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/803192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/803192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/803192/1"}}},"commit":{"parents":[{"commit":"036948c516672e55c4a2c932c74bcdec80319108","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/036948c516672e55c4a2c932c74bcdec80319108"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 06:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 11:14:15.000000000","tz":720},"subject":"Use OVN as the default Neutron backend in DevStack","message":"Use OVN as the default Neutron backend in DevStack\n\nChange-Id: Ie995eb0c5ddb6fcc0d400a8200a204a026afd173\n\nUse OVN as the default Neutron backend in DevStack\n\nChange-Id: Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b83494c8d32506b1db88aebaa72c2330d08b383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b83494c8d32506b1db88aebaa72c2330d08b383"}]},"branch":"refs/heads/master"},"a94046e3b092e48f3c923846432dc862abd0ee30":{"kind":"REWORK","_number":2,"created":"2021-08-04 00:55:08.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/803192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/803192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/803192/2"}}},"commit":{"parents":[{"commit":"036948c516672e55c4a2c932c74bcdec80319108","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/036948c516672e55c4a2c932c74bcdec80319108"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 06:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-04 00:54:46.000000000","tz":720},"subject":"Use OVN as the default Neutron backend in DevStack","message":"Use OVN as the default Neutron backend in DevStack\n\nChange-Id: Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a94046e3b092e48f3c923846432dc862abd0ee30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a94046e3b092e48f3c923846432dc862abd0ee30"}]},"branch":"refs/heads/master"},"cc88f1da40eabfd794b997a666d414f6f5b89d0d":{"kind":"REWORK","_number":3,"created":"2021-08-05 23:11:22.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/803192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/803192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/803192/3"}}},"commit":{"parents":[{"commit":"036948c516672e55c4a2c932c74bcdec80319108","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/036948c516672e55c4a2c932c74bcdec80319108"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 06:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-05 23:11:08.000000000","tz":720},"subject":"Use OVN as the default Neutron backend in DevStack","message":"Use OVN as the default Neutron backend in DevStack\n\nChange-Id: Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc88f1da40eabfd794b997a666d414f6f5b89d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc88f1da40eabfd794b997a666d414f6f5b89d0d"}]},"branch":"refs/heads/master"},"18371465fcb22ce4aae993e487e9a620e8959ff2":{"kind":"REWORK","_number":4,"created":"2021-08-16 00:47:59.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/803192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/803192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/803192/4"}}},"commit":{"parents":[{"commit":"036948c516672e55c4a2c932c74bcdec80319108","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/036948c516672e55c4a2c932c74bcdec80319108"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 06:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-16 00:47:47.000000000","tz":720},"subject":"Use OVN as the default Neutron backend in DevStack","message":"Use OVN as the default Neutron backend in DevStack\n\nChange-Id: Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/18371465fcb22ce4aae993e487e9a620e8959ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/18371465fcb22ce4aae993e487e9a620e8959ff2"}]},"branch":"refs/heads/master"},"3638e686884d4ae14a63424274e6f1595f5fa8af":{"kind":"REWORK","_number":5,"created":"2021-10-09 19:19:32.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/803192/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/803192/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/803192/5"}}},"commit":{"parents":[{"commit":"be86b79119d16ee77f596172f43b0c97cb2617bd","subject":"[Doc] Describe the usage of root_on_create option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be86b79119d16ee77f596172f43b0c97cb2617bd"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 06:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-10-09 19:19:12.000000000","tz":780},"subject":"Use OVN as the default Neutron backend in DevStack","message":"Use OVN as the default Neutron backend in DevStack\n\nChange-Id: Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3638e686884d4ae14a63424274e6f1595f5fa8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3638e686884d4ae14a63424274e6f1595f5fa8af"}]},"branch":"refs/heads/master"},"7997fe91a9161a58816f65e2663615cd230a7074":{"kind":"REWORK","_number":6,"created":"2021-10-10 01:22:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/803192/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/803192/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/803192/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/803192/6"}}},"commit":{"parents":[{"commit":"be86b79119d16ee77f596172f43b0c97cb2617bd","subject":"[Doc] Describe the usage of root_on_create option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be86b79119d16ee77f596172f43b0c97cb2617bd"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-02 06:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-10-10 01:22:07.000000000","tz":780},"subject":"Use OVN as the default Neutron backend in DevStack","message":"Use OVN as the default Neutron backend in DevStack\n\nChange-Id: Ia7d08b9b7ef0c59f6a0284a506949ebb4d2ccf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7997fe91a9161a58816f65e2663615cd230a7074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7997fe91a9161a58816f65e2663615cd230a7074"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
