)]}'
{"id":"openstack%2Ftrove~805941","triplet_id":"openstack%2Ftrove~master~Iebac740e982a89fcf882a2cfc3e447ac53ee6656","project":"openstack/trove","branch":"master","topic":"fix-show-user-quota","hashtags":[],"change_id":"Iebac740e982a89fcf882a2cfc3e447ac53ee6656","subject":"Allow regular user to get quotas","status":"MERGED","created":"2021-08-25 05:15:27.000000000","updated":"2021-08-27 04:15:59.000000000","submitted":"2021-08-27 04:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805941-fix-show-user-quota","meta_rev_id":"a5baeea2af91793cb6e11c74ed465faafcf09b43","_number":805941,"virtual_id_number":805941,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-27 04:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-08-27 02:53:43.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-08-27 02:53:43.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 07:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 07:52:53.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"state":"REVIEWER"},{"updated":"2021-08-25 07:53:05.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"state":"REVIEWER"}],"messages":[{"id":"20b139fe47bf45af12f97526d7c4799859194635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-25 05:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a79456e97ff988abb1045b044d2a120124f39ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 07:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/79d69244aade4f70b4899c3f563801fa : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1b501fbb22e418387c428c2933afc59 : FAILURE in 8m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b63e621d874e4fb7462b1841d40ad0 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8ec283f17de49fe9c3c9a75a88d20a5 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afde799692f14cd59feb8120c0e3719c : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01f8f15de0844b47b0a38319ee24c7a4 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df59fed2e4f3411087b03ab17e78c610 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/570e46c129934f8ebb251fdc4a1da1bc : SUCCESS in 7m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b0573d4c7e44bdb94e0f4afb5a1fecb : SUCCESS in 7m 46s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1aa9ccb779e44bac92b17d6a6a293268 : FAILURE in 46m 37s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/43a50c99a1fe4f08aa525a5b80dee127 : SUCCESS in 1h 21m 14s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8a84c1a0afe47f3be7ab3bfda4642ce : FAILURE in 1h 49m 45s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2a4e67dd769f4c66a646a9d686c9067b : FAILURE in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a74ebb82612142249858f45b4861baca55c87da","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-27 02:53:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29976a8f88b33dc7e1c1e0b33fa0a0f0af4c4e17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 02:54:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffaa51bd1374742ef317b1a33fda1bd0655aaf90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 04:14:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b90dfd5d9bcd4343ab5d4903b24dc16e : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f55417e349f2425da114256fd18c42a5 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d3c7dd5d16943c586b31127d34efbb0 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f608d3fdc0f4b09ac3d027bdb729489 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc9f12ed3b94aa7aa80d38ff0a87600 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc66cbd2313149509ee954e56195b6c8 : SUCCESS in 6m 40s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/856d0c31635143d1b9cf0f4758dbc997 : FAILURE in 1h 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da8b81181948060b2f4cd74b92f476fd872f6e46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 04:14:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a5baeea2af91793cb6e11c74ed465faafcf09b43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 04:15:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ea7fd3fcb9324da781f93c6360d08a33 : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c49c4bc8f3b4f71964acd6252ec3c80 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2196ecbd1b3040fdba4338f239b03b19 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3459e2662e86b275ac3baf01330753df4d3f092","revisions":{"f3459e2662e86b275ac3baf01330753df4d3f092":{"kind":"REWORK","_number":1,"created":"2021-08-25 05:15:27.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/41/805941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/41/805941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/805941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/805941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/805941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/41/805941/1"}}},"commit":{"parents":[{"commit":"c3a8930eb152072ab71ccc6033862b3ec45f5e4f","subject":"Doc: Upgrade trove guest agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c3a8930eb152072ab71ccc6033862b3ec45f5e4f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-25 04:11:12.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-08-25 05:14:42.000000000","tz":720},"subject":"Allow regular user to get quotas","message":"Allow regular user to get quotas\n\nThe project user can query the project\u0027s own resource quota.\n\nStory: 2009140\nTask: 43082\n\nChange-Id: Iebac740e982a89fcf882a2cfc3e447ac53ee6656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3459e2662e86b275ac3baf01330753df4d3f092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3459e2662e86b275ac3baf01330753df4d3f092"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
