)]}'
{"id":"openstack%2Ftrove~806295","triplet_id":"openstack%2Ftrove~master~I3929511a621790c504cba16be17933211c26107a","project":"openstack/trove","branch":"master","topic":"add_user_client","hashtags":[],"change_id":"I3929511a621790c504cba16be17933211c26107a","subject":"Support port was created from user\u0027s network should visible with user","status":"ABANDONED","created":"2021-08-27 02:24:28.000000000","updated":"2021-08-27 03:36:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9791657407e8de8210c64b3ac1a649c619e940a","_number":806295,"virtual_id_number":806295,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-27 03:18:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-27 03:18:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eca9e6e2a4a6032346d87a628ccddf144976a794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-08-27 02:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bcadd69d0ba397f229bc92a73f1db91dce00622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 03:18:07.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e014fa72afbe4546b03869ce47ee777a : FAILURE in 7m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec589a228386481c8dac1f18aab83905 : FAILURE in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d6ef432100845a480625b617967e6ff : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/023a995b5cfa4d45aaf635e670b349a4 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f520c0e528b4decb19c5fae46848eb3 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65dda6f9f01c4aa5bf0b40fee5717c6a : SUCCESS in 7m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/956377bdc0ba4b409896feb96ee75e4d : SUCCESS in 7m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e7448a22f8f940cf876dcac044fff249 : FAILURE in 33m 55s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/dfb06029eecb49a58cd3056eda9b176e : FAILURE in 38m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44f0fe9f6215471db1c17f2477f23a4d : FAILURE in 48m 19s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d5d114096a4d479ca349aeac8d50a059 : FAILURE in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9791657407e8de8210c64b3ac1a649c619e940a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-08-27 03:36:49.000000000","message":"Abandoned\n\nNova not support.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c88cc5507f84373ab00b0eb66849756b779cb77","revisions":{"0c88cc5507f84373ab00b0eb66849756b779cb77":{"kind":"REWORK","_number":1,"created":"2021-08-27 02:24:28.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/95/806295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/95/806295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/806295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/806295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/806295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/95/806295/1"}}},"commit":{"parents":[{"commit":"c3a8930eb152072ab71ccc6033862b3ec45f5e4f","subject":"Doc: Upgrade trove guest agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c3a8930eb152072ab71ccc6033862b3ec45f5e4f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-08-27 02:20:46.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-08-27 02:20:46.000000000","tz":420},"subject":"Support port was created from user\u0027s network should visible with user","message":"Support port was created from user\u0027s network should visible with user\n\nTask: #43101\nStory: #2009147\nChange-Id: I3929511a621790c504cba16be17933211c26107a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c88cc5507f84373ab00b0eb66849756b779cb77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c88cc5507f84373ab00b0eb66849756b779cb77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
