)]}'
{"id":"openstack%2Ftrove~807473","triplet_id":"openstack%2Ftrove~master~I469ad8b44640118883c903accfef0349e9ecb880","project":"openstack/trove","branch":"master","topic":"rename_variable","hashtags":[],"change_id":"I469ad8b44640118883c903accfef0349e9ecb880","subject":"Make more transparence with variable name from replica snapshot","status":"MERGED","created":"2021-09-05 06:41:59.000000000","updated":"2021-09-30 22:44:28.000000000","submitted":"2021-09-30 22:43:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807473-rename_variable","meta_rev_id":"5ba9772dc2d3f8f8279db4b1a4a562572508d04f","_number":807473,"virtual_id_number":807473,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 22:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-09-30 21:07:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-09-30 21:07:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-05 08:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 21:07:52.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"f758bb55822efe797d6040de65e6f83b7391f946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-09-05 06:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08876717d86eaaf5ff495c0aea7e3987e43ccd9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 08:31:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f72a87d7474d48dbb926ef53db08d04f : FAILURE in 9m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2c1f84ec664d29b82555969a3060fa : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e12a98aebee467ab1a26adf73517785 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9987878c38a9427c98999f61cb54a470 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/445cf70fc58f4744840d6f7b4bc84e18 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f3ca71516a4606afe827482d99ac0a : SUCCESS in 8m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/263f2f5ed3c543a68d749b6c1e97db48 : SUCCESS in 9m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2b6d1dd730bd49ef81ce51385053d90f : FAILURE in 1h 23m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/229acbe7ac094b52ba8b9118b7aa56a2 : FAILURE in 1h 47m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce5042f31cd9463381d286f47742ec63 : FAILURE in 1h 40m 34s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/69d485abeaeb493b9795149fe6907cbb : FAILURE in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48f1284ef9296a9fa48b3a4e0079eb02e5b9090d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-09-30 21:07:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f10caeb1ed7f49760253d73fe4330c56098e7db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 21:09:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e45b35bb307bce9adab852a776be3ba0155ebad5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 22:43:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f1aac447a4e4bdabfbd19d7d506a55a : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cd3ed6252ce46ffbc109835412ae705 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/080a34461e444ae58cf6fc59d9200915 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f878f85cbba74889bc7ccea01ad6f4b4 : SUCCESS in 6m 28s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a029ab557f83454f9fc906d1a650cac8 : SUCCESS in 1h 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e127d601a192d17b267624bd8f3e80348714dfbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 22:43:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ba9772dc2d3f8f8279db4b1a4a562572508d04f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 22:44:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c36f498c2c0441197f169971b0b765e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edd503976c6a502f8236c69c3be8ab2b5c39aaf7","revisions":{"edd503976c6a502f8236c69c3be8ab2b5c39aaf7":{"kind":"REWORK","_number":1,"created":"2021-09-05 06:41:59.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/73/807473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/807473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/807473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/807473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/807473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/807473/1"}}},"commit":{"parents":[{"commit":"f3459e2662e86b275ac3baf01330753df4d3f092","subject":"Allow regular user to get quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3459e2662e86b275ac3baf01330753df4d3f092"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-09-05 06:41:31.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-09-05 06:41:31.000000000","tz":420},"subject":"Make more transparence with variable name from replica snapshot","message":"Make more transparence with variable name from replica snapshot\n\nTask: #43152\nStory: #2009162\nChange-Id: I469ad8b44640118883c903accfef0349e9ecb880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edd503976c6a502f8236c69c3be8ab2b5c39aaf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edd503976c6a502f8236c69c3be8ab2b5c39aaf7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
