)]}'
{"id":"openstack%2Ftrove~807474","triplet_id":"openstack%2Ftrove~master~I6aff0d185991db77b17a4f23084de6734b2b3a6c","project":"openstack/trove","branch":"master","topic":"improve_parse_log","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2022-02-13 11:25:48.000000000","reason":"Change was submitted"},"33822":{"account":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"last_update":"2022-02-13 10:48:37.000000000","reason":"Qiannan Lyu replied on the change"}},"hashtags":[],"change_id":"I6aff0d185991db77b17a4f23084de6734b2b3a6c","subject":"Improve the detect backup was created when parse log","status":"MERGED","created":"2021-09-05 06:54:16.000000000","updated":"2022-02-13 11:27:43.000000000","submitted":"2022-02-13 11:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807474-improve_parse_log","meta_rev_id":"bfa6c9f69cda25fe0bf456a2a9ad19546021b9e0","_number":807474,"virtual_id_number":807474,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-13 11:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2022-02-13 11:00:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-13 10:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2022-02-13 11:00:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-05 08:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 20:38:50.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2022-02-03 03:00:23.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"state":"REVIEWER"}],"messages":[{"id":"a36e56e281604bc16c3dbfe75b951efeb5af3d48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-09-05 06:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d57f70e6cdef425609258ee660f8dd7eff44aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 08:38:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/663775a6a8494e66836d8237c85c35ca : FAILURE in 7m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979cd9fe92444534b0320a9327101d55 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd1290f7c40a491fa9bce054385e0488 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88725902e2bd4c789aebbd8853d23362 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d48a4d310154c63b0ef4e50b9b30bf6 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f01818defe4b7a91b28b5624abf559 : SUCCESS in 6m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb825968bf8e4c32a824d765080bfddb : SUCCESS in 6m 25s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bd412c1b74a9407c8300353bb848f4d6 : FAILURE in 1h 42m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/948ddc3370d54824b892000aec9f8afe : FAILURE in 1h 21m 22s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf3dc2e1f70a46a8bbd4060be0e0d5a7 : FAILURE in 1h 21m 47s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9097a7c9fa82432bbedf4f34c3dfdb25 : FAILURE in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0d35ab7d712f6d1fc86e6bf977c561d53814b8d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-09-30 20:38:50.000000000","message":"Patch Set 1: Code-Review-1\n\nWe should move this line(https://github.com/openstack/trove/blob/master/backup/main.py#L115) outside of the `with` section.","accounts_in_message":[],"_revision_number":1},{"id":"b3a44b6487bc4a3b543fd8980ed3be73ea0f39bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-01-21 03:32:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a26f19df811e2b626eca2f7bb3d33e8f0a1568ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-01-21 03:34:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5e1a5d25da3f5c0b06c48509c31146924553f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 04:28:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cafa94ebc1d74ee78093cf0c064de122 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1621f391a1a04b8aaaa2d96a8bff7faf : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/682d23abf0a94206a904942509e0f05b : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d117fcb4bc841dfbef67300396a6b3e : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ebb0caa97404ee386a9edd05c0efcd4 : SUCCESS in 6m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2cda35e414574f2baf93ca1cae0cba6e : SUCCESS in 8m 35s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d45a1485fb6b4c3aad48aa28868127a3 : FAILURE in 27m 14s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/910355c4f758483d9d90b55146e66b48 : FAILURE in 27m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34f7d27852414b2b8a4c0bff780316bb : FAILURE in 37m 03s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/58b3271131eb4770a06a78d9e53e9f48 : FAILURE in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27bb551c730b0ef1187932cd3eaf66aee114113f","author":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"date":"2022-02-13 10:48:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"08bb2fa1062eca0080fe062941176a99fde3e7df","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2022-02-13 11:00:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"131269e964bff896ffdc0bc629cc331a6ecb05db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:00:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e6c55c5098fb4e223a053b4fab76e20d534f872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:25:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b693e0c0ef408394883b753cca6656 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd7c0b45d754666b61c068461a94230 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e000366255314e6097645a2615a96ced : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad9005e9a874f2890bce66416e62dd6 : SUCCESS in 5m 01s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/39c78d90d8ca437dbb62ae8e88e735aa : FAILURE in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4756f303892e56b62c1c7c7512e482ae482727b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:25:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bfa6c9f69cda25fe0bf456a2a9ad19546021b9e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:27:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4baad5f36d249f99aec53b639bc0776 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"91e002b1f2a509407055590e79a6ac489ab90755","revisions":{"b0f8014672db5830609b6ad07c7662ad1707fabb":{"kind":"REWORK","_number":1,"created":"2021-09-05 06:54:16.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/74/807474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/807474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/807474/1"}}},"commit":{"parents":[{"commit":"f3459e2662e86b275ac3baf01330753df4d3f092","subject":"Allow regular user to get quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3459e2662e86b275ac3baf01330753df4d3f092"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-09-05 06:53:24.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-09-05 06:53:24.000000000","tz":420},"subject":"Improve the detect backup was created when parse log","message":"Improve the detect backup was created when parse log\n\nTask: #43209\nStory: #2009176\nChange-Id: I6aff0d185991db77b17a4f23084de6734b2b3a6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0f8014672db5830609b6ad07c7662ad1707fabb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0f8014672db5830609b6ad07c7662ad1707fabb"}]},"branch":"refs/heads/master"},"099ab05c25fd99b4b331aa3a29b1f7f572870d5c":{"kind":"REWORK","_number":2,"created":"2022-01-21 03:32:49.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/74/807474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/807474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/807474/2"}}},"commit":{"parents":[{"commit":"f3459e2662e86b275ac3baf01330753df4d3f092","subject":"Allow regular user to get quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3459e2662e86b275ac3baf01330753df4d3f092"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-09-05 06:53:24.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-21 03:32:30.000000000","tz":420},"subject":"Improve the detect backup was created when parse log","message":"Improve the detect backup was created when parse log\n\nTask: #43209\nStory: #2009176\nChange-Id: I6aff0d185991db77b17a4f23084de6734b2b3a6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/099ab05c25fd99b4b331aa3a29b1f7f572870d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/099ab05c25fd99b4b331aa3a29b1f7f572870d5c"}]},"branch":"refs/heads/master"},"91e002b1f2a509407055590e79a6ac489ab90755":{"kind":"REWORK","_number":3,"created":"2022-01-21 03:34:48.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/74/807474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/807474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/807474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/807474/3"}}},"commit":{"parents":[{"commit":"f3459e2662e86b275ac3baf01330753df4d3f092","subject":"Allow regular user to get quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3459e2662e86b275ac3baf01330753df4d3f092"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-09-05 06:53:24.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-21 03:34:32.000000000","tz":420},"subject":"Improve the detect backup was created when parse log","message":"Improve the detect backup was created when parse log\n\nTask: #43209\nStory: #2009176\nChange-Id: I6aff0d185991db77b17a4f23084de6734b2b3a6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91e002b1f2a509407055590e79a6ac489ab90755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91e002b1f2a509407055590e79a6ac489ab90755"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
