)]}'
{"id":"openstack%2Ftrove~809109","triplet_id":"openstack%2Ftrove~master~I83cdddcccd1340de950e4dd3a5ab374553cecad1","project":"openstack/trove","branch":"master","topic":"add-yoga-python-jobtemplates","hashtags":[],"change_id":"I83cdddcccd1340de950e4dd3a5ab374553cecad1","subject":"Add Python3 yoga unit tests","status":"MERGED","created":"2021-09-15 02:34:45.000000000","updated":"2021-09-15 07:55:29.000000000","submitted":"2021-09-15 07:45:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809109-add-yoga-python-jobtemplates","meta_rev_id":"e749ab7802887a1cf2aa02464fa0452a6634f5da","_number":809109,"virtual_id_number":809109,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 07:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-09-15 05:49:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-09-15 05:49:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 03:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 05:49:45.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"80306b841a3cdf1ce22da141bf339894e3ca3e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-15 02:34:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d66c27596f0ab245b422f1c1ee2bd1d03426ac1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 03:34:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac415ef8b9a44ae6bad159d6216c6895 : FAILURE in 9m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1fe9f49358f4b5caf98bbea5e4cad23 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e346b3861b940f9bd49e621224f0f5a : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d5e4629f41f4905809745d02f8b8844 : SUCCESS in 9m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e22809df3314420a53bb6f26508ac0b : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49088f4c3e384d8d9c928543ffb7549d : SUCCESS in 8m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e4c20a571ef41f1ab4e141c591854db : SUCCESS in 8m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/76cdf9e2c86b48988b4affb4b6e05cf7 : FAILURE in 46m 16s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/23e3fb053ac84334ad2ab0ebabc566d2 : FAILURE in 34m 19s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/003a170bf3184d46b2ea53fc8516a04d : FAILURE in 48m 56s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0cd5873950384a62ab4843503f0418a0 : FAILURE in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f1aef923e7a787b4f74afb2df0e79434c69cbff","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-09-15 05:49:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4312ea30019df101a96ec757b0db6c702967c98e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 05:50:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3020c9423b7b27c3995e00b76dc118fea7b477d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:45:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c780b67f55472da789c92480adf24c : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b63a5d9d0740fcac1fda7c08b06279 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c7ec7331f8643d5abc299bb2c991861 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb2b0058ca14091a81aec3bb758a558 : SUCCESS in 6m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1cfba85cf1e242f6a6c70cb52efc278e : FAILURE in 1h 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"701e18e0383a302a3f94c1f9ad10c5f4b194ec65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:45:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e749ab7802887a1cf2aa02464fa0452a6634f5da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:55:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d1349e5daf49eb91c3bd6c7c0ba625 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2f1e8bdb6ae4ba1360584ae870310c92756bb1f","revisions":{"c2f1e8bdb6ae4ba1360584ae870310c92756bb1f":{"kind":"REWORK","_number":1,"created":"2021-09-15 02:34:45.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/09/809109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/809109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/809109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/809109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/809109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/809109/1"}}},"commit":{"parents":[{"commit":"b9f77ef965d6dc1bd4b325cb1fd9235b576e9155","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9f77ef965d6dc1bd4b325cb1fd9235b576e9155"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:34:44.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:34:44.000000000","tz":0},"subject":"Add Python3 yoga unit tests","message":"Add Python3 yoga unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for yoga.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I83cdddcccd1340de950e4dd3a5ab374553cecad1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c2f1e8bdb6ae4ba1360584ae870310c92756bb1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c2f1e8bdb6ae4ba1360584ae870310c92756bb1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
