)]}'
{"id":"openstack%2Ftrove~812016","triplet_id":"openstack%2Ftrove~stable%2Fxena~I669fedc76aa79d7bfc1e4e64684adc4a89a3a15d","project":"openstack/trove","branch":"stable/xena","topic":"guest-agent-log-stable/xena","hashtags":[],"change_id":"I669fedc76aa79d7bfc1e4e64684adc4a89a3a15d","subject":"Fix trove guest agent logrotate","status":"MERGED","created":"2021-09-30 20:54:09.000000000","updated":"2021-10-01 03:41:19.000000000","submitted":"2021-10-01 03:40:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812016-guest-agent-log-stable/xena","cherry_pick_of_change":811938,"cherry_pick_of_patch_set":1,"meta_rev_id":"5ba84557d3622a28a46013c894e69f58f8a6f2c0","_number":812016,"virtual_id_number":812016,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-01 03:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-10-01 00:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-10-01 00:09:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-30 23:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a91b50f7e7a9b5c5e1adaa664dcf849b6f8548de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-09-30 20:54:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"217e09ca48883686dae51154364d194bea8a3b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 23:00:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c5f79861f34f68bca4d2a6afafee2b : FAILURE in 6m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f1f76b8b934ad58db225df43bcaa6a : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d59161314ed47f7a8ec9c75a4d13188 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/084b59a593484483995cd794d725276f : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/201e8cb65ba14965891176128d67d5c2 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8978ea864b97415f857e606317f63ccd : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd0ab8fcc9af4754bf2823d92ee085f9 : SUCCESS in 5m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/90160d2232064642a2e8b30ca6b722ed : SUCCESS in 1h 36m 44s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6b78f4bb786042fd83a36ddd4f8244c0 : SUCCESS in 1h 20m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48fd1933aed94106b6be134119466176 : TIMED_OUT in 2h 03m 46s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8d6c3c7c3927458096efa7b92f8bf8cc : FAILURE in 32m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbc2f466f6d9f205e8b1904e15b2f99cd0125d35","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-10-01 00:09:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29dfd760e23ff3b49d25ad912859f3979a6aa9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 00:10:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ab36a3d13d44889b7e0e337f4fe9a378f57d5e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 03:40:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6fbfcebeebd416bb192c49a3f4da8ee : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45cd1b4542744c3480bc677cefb76642 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2deafc3bc71b452a82defcbc931a4f5d : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dec9e81c57f42699ca8d51f6a0b4111 : SUCCESS in 7m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0694d69029c2448aa3d644b8c4328813 : SUCCESS in 1h 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e2aeccfceaea23b57aaa038b822c19c8b5e5d10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 03:40:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ba84557d3622a28a46013c894e69f58f8a6f2c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 03:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e503db3ec28341268c2a6f0fcbf68153 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3864fc262121881bd676f832c363b9ae3ea9c2d","revisions":{"f3864fc262121881bd676f832c363b9ae3ea9c2d":{"kind":"REWORK","_number":1,"created":"2021-09-30 20:54:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/812016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/812016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/812016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/812016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/812016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/812016/1"}}},"commit":{"parents":[{"commit":"42df225160ad712b1cf53ce98ee4ab9c5938d286","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/42df225160ad712b1cf53ce98ee4ab9c5938d286"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-09-30 09:27:53.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-09-30 20:54:09.000000000","tz":0},"subject":"Fix trove guest agent logrotate","message":"Fix trove guest agent logrotate\n\n* The logrotate file owner should be root\n* Log file path is /var/log/trove/trove-guestagent.log\n\nChange-Id: I669fedc76aa79d7bfc1e4e64684adc4a89a3a15d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3864fc262121881bd676f832c363b9ae3ea9c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3864fc262121881bd676f832c363b9ae3ea9c2d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
