)]}'
{"id":"openstack%2Ftrove~823164","triplet_id":"openstack%2Ftrove~master~Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0","project":"openstack/trove","branch":"master","topic":"support_root_volume","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-12-15 08:20:43.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-12-15 08:20:43.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-08 02:09:13.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"33822":{"account":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"last_update":"2023-12-15 08:20:43.000000000","reason":"Change was submitted"},"33551":{"account":{"_account_id":33551,"name":"Ammad Ali","email":"syedammad83@gmail.com"},"last_update":"2023-12-15 08:20:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0","subject":"Cinder Support For Boot Volume","status":"MERGED","created":"2021-12-29 16:01:57.000000000","updated":"2023-12-15 08:21:54.000000000","submitted":"2023-12-15 08:20:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":64,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"823164-support_root_volume","meta_rev_id":"86da08f38eb2bfa84f4ede8aa50298148123b358","_number":823164,"virtual_id_number":823164,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-15 08:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-14 01:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-14 10:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"CC":[{"_account_id":33551,"name":"Ammad Ali","email":"syedammad83@gmail.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-29 16:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 03:00:46.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"CC"},{"updated":"2022-02-03 03:00:46.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"state":"REVIEWER"},{"updated":"2022-06-20 12:17:00.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2022-06-20 12:17:00.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2023-03-07 09:57:42.000000000","updated_by":{"_account_id":33551,"name":"Ammad Ali","email":"syedammad83@gmail.com"},"reviewer":{"_account_id":33551,"name":"Ammad Ali","email":"syedammad83@gmail.com"},"state":"CC"},{"updated":"2023-03-08 09:51:24.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REMOVED"},{"updated":"2023-07-05 09:51:23.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-11-13 09:35:27.000000000","updated_by":{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"},"reviewer":{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"},"state":"CC"},{"updated":"2023-11-13 09:35:51.000000000","updated_by":{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"},"reviewer":{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"},"state":"REMOVED"}],"messages":[{"id":"c7cac78b92498d79c11d3a2ad8c9f548bee830a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-12-29 16:01:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df7fc6711ff053fef82e4ceb84bf447a240b268b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 16:36:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e528003d9d64504aa12ad5e6ffc101d : FAILURE in 8m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daccfffe4f4d49b59028eea2eccbb385 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15995cd184cc4123829fc1298b42c7aa : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98d2278e8c664d70b20b347386c877c0 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31019915731a4ffba232700be3dc2a31 : SUCCESS in 7m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c054d71cdb7e4733a3066ba146fa7617 : SUCCESS in 6m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/df644ed20ad24f3b96a3041a7e51d7e7 : FAILURE in 32m 10s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/85e68a4f339047c88c7dcc33f5bffeb3 : FAILURE in 29m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4169b69a228d4f1fa7580ff762606402 : FAILURE in 30m 17s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b24c39f776e94d3fb6c3f5431991c250 : FAILURE in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e2a9f2113c136664179bd1cd65d7257f055d019","author":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"date":"2022-03-29 21:00:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c68221c2f9c3de9103f90dedd99bd5cf2dc61ad8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-20 12:17:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f4e2dc29002168aad442a2b772351ab82c27687","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-06-20 23:07:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"00031d0f10826beb331a6b89e3075ea7e8aa459a","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-07-12 09:56:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66ac5de3bae901fa83eda66f763944d8191152b8","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-18 08:00:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de199e6a44164127ae6b86beaf25cc3712bdf4c5","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-02-17 03:19:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"847292b65386404a19f71179b3ce1e375fd5362a","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-02-27 01:56:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"788fd195382c06e783fcf900a646970562b0441b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-02-27 02:11:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"496adad192cfc1f8197f824a48061fe040774083","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-02-27 05:22:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfd41f114926004a02c8d092899dd2ab9739df13","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-02-27 07:17:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"375abf4f2e42357d75c5ecc8cc44aef8b11de64d","author":{"_account_id":33551,"name":"Ammad Ali","email":"syedammad83@gmail.com"},"date":"2023-03-07 09:57:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80f01c651de62db7248111a38afd1242a139c0d3","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2023-03-08 09:51:24.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_6732\u003e.","accounts_in_message":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"_revision_number":1},{"id":"45ae3969a269c833a9b36d29588187562ab99051","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-28 12:08:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15d7b50138ba2a9e038411c72f92bb9d81f5ea50","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-05 02:59:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbc4bd1c1d7c5ca0f68adcf7b95e5f7700701a82","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-05 09:51:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1023baaad6723f778082bd619fb16c42e99b8ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-02 08:12:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b5fac70ea65589a637a8ccf1a0db353d0aa9a17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 08:22:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85b15d2ef47e4012945803844039b183\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b69334c79384284bcc50bec7b130802 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afa876f552ea44d09e4c9462e52dc99c : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c93bfd9efa543bd8b98cc85e865f38a : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c447e06b54dec7d23ce606138f8e541cdece0a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 10:54:43.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f44f1136dddb407aab2ce8e680395135\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2318897587a943a4b996dce85b881771 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6ac856e7f64d9883a7be34347fe408 : FAILURE in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d02eea60364dfbad19a84659fae70d : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0f57db546e7411285c03cfc5dfc3564 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d79855d22f5b4ec59d329024db67bdda : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b599f77ac59d41f6ad0a9ad5c95f5dc0 : SUCCESS in 6m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eea86ef474bf4f9e9db40b0006b5001a : SUCCESS in 4m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0c50032e58d34bc4bdc3a3b5393d3aa1 : FAILURE in 2h 40m 21s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1ff8a6e7335a4a2b9f3b66000d4022a2 : FAILURE in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69bc69019e4765282c90643f83b28021b24a29d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-03 02:00:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"67525f7a3df035c6bd146a2f6af9c1e7e418ad2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 02:07:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f28011e9594e6dbd746728e048dabc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0729ad9697f445e080e5c101d09c5b90 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfa43efcc7d04ab9a8e169f9ee1d6bde : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c4d652f6d2040938c123b97dcb8c136 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"918fc8e7d2c16f3a89e25b029aa0ca72bb71fcdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 04:55:51.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18da92d842d8498c8e7c8bb43aa5db0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d17fdbb25fda4559abebdffcc058109d : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca84e1ab83e468f94280e96d9da4136 : FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1055e41f26549b487e876421176b7fb : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/119357cc5d774c719d2c15a18c027fd3 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f95399d4acf44007a07f234f345b5416 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd2a05c9a904fc594966473f7e040aa : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b91d6c82b7e4f479183ac3b4620d87c : SUCCESS in 5m 35s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/145d3fed0d85401c94e63363c1cfd96f : FAILURE in 2h 53m 15s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/f7dbe2f5b45341169fb7c8fee6ef2fc0 : FAILURE in 2h 06m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a47c3790ed1bfe05c335bc54f858d32b2e64287e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-03 06:43:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ea5eb255e8a296c3f857c0873347ad5293ec8fb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 06:49:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9f853bc349482faf01139912d1f003\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48b4cfe2680549f9b81e0e02343b9168 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d75b374ef4845d4a0c672dd3729cd70 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/994f3fa1d858431e88047b7e5aea13d0 : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9081e0652c310f7849fb255a7fc245c96f559693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 09:08:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec46a89824e4c308ed232c43507d4c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d827a1a31704d9f830a6d803bc519bd : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6565179e8f84455b85e54230505f60a : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cae1f0f6a7934c41acd2ba21ba4454b1 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a0192d9757d4a9b9ff762b90f019e4d : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bab89fb18ea5489087c35339b51680c0 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e31bf6245144a68aaacc6bd90b486f : SUCCESS in 5m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b32afd401ea6476990878e658f170e8b : SUCCESS in 5m 18s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d90b7d21c8a94c09afa87c1a01ace36f : FAILURE in 2h 00m 04s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ea40e62b431947b5a8907e4b6a34a61f : FAILURE in 2h 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b2fdc7ae061298cf2b09d5cd8d49186b5c1793f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-07 04:51:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d23d7e61945ec4edb41c36c616ad71cd2b9fa791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-03 04:26:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2bf7e01b4ab291bdb46c7bf6b7dec1214251b8fe","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-03 04:29:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4eb95d0cc677b518dff93c082916a8487ecc1b6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 04:41:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5160755436ee4003b6d81dc48f1cbe48\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0acd1eaa88044f7a1d9750e11132cbb : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8aa66349967c4542929c1231d8226e3f : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/836c6abdf122444294f0f8ddec007989 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa1515bd8fa1e0b2e24919362e574c54ea5f830c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-03 06:10:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8ad0edc3e393506d5241ac677c1238f5e8b9ad74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 06:48:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f52e3df831854c51a3afba309e397404\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b9a36765c56420b9c20228af0a6e3ce : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ec1bf1f0a149669bb2caf2e3703ef3 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/878a99fb76854fc699becd758df4bd94 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0c7e17945c74ae9ad16d2da6ebae493 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78afe960602441da55d4d926141bee8 : SUCCESS in 5m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b940063e66c2452f8d2af0e6a43d0f2b : SUCCESS in 5m 19s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c96d8f55ab49486bbb1325c22dbfe588 : FAILURE in 2h 19m 24s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/1c4d9bb8c84b4f918809e597281ecafb : FAILURE in 1h 12m 05s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/14f0e17f08884d4da5fda5d15cd3f45e : FAILURE in 1h 55m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"27814474d2d383f084a672b30199295545119d0c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-03 07:05:43.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"db39c45e9e9b9dcd1d7ee96e0ba9e89387bef1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-03 07:06:57.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7a1af03614bb77f6319b2110a91b7ad7ed3e8eb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 07:19:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/077eb7e13307482d8d0d066de8917136\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e13792faaf34253abda634e832eae96 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/feecd6717fb0483aa0992ec47a07d5b8 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b7bbde683a74a1897ebdaba52b5336b : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9bf4fd11c75c8e2023622b5c9922bd82cd1863ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 08:46:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dc4eccf610e4e5395aa479dcb433da7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/034c6de0afe54bbfa0d7b7356cc9cc6e : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11607ccfe834a5c89a58bf0e8e3eaa9 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b17771d14e0f45d1aebc757de22cf89d : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f15ee0ba5b194911872f2db85616e3ac : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b292982741364d4d8081e13816b9a3a9 : SUCCESS in 5m 33s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/19a285c28c124a27829f731a30e0da52 : FAILURE in 1h 35m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/026b270787fe4148bdd466f281550935 : FAILURE in 1h 18m 56s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d5f88c520d864cefa4708f9155a3a65c : FAILURE in 1h 18m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/230ad4a41dc64e6d9a02bfd7e8d3b08b : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":6},{"id":"1c4cbe719c391813778f8f045439b512fad199a4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-03 08:55:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"35af2dab67e6c7ab0140565ce5eba359bf20d44b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-06 03:21:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"057d6fff92d4fff2253f3bdfb0f3c208291f8553","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-06 03:26:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b589ee11b50703450bf70a15fbec5c2593fd3279","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-07 08:04:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ee31e5075b5416a8915eb51511d212aada0c607e","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-07 08:32:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"61127a17cfd24bd0c995d099db866dc22f42f27a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-07 09:34:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0f2d1aa7a82fb65df569096ad6eb6862654a0889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-08 04:28:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d2795f225dec8e7a96fff30939d6e34f9a9ff8cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 04:46:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b35f95b273ac43ceb6bccfc6a8c25fdb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d90d0f4a8e140549c81a70b163fc3b7 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7425789c863414c80654fcee583aa4a : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/002e1925ddbe48f7aafe1e31c99a9007 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"138c25d9c0bce9b87b429e18bd283bb94c3d7a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 06:22:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7936da303262471baee87f4ec9848612\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d63d79d98b56448cbcef1f397c9a990b : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1099a928fdcb446285a509e23a84cd32 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef66c8498f25492aa5ac3037255457f8 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28a0999b1e3149e19f0956beccecda41 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be45a58cc5d48e792c6f4801f35991d : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8962fce998549c680d2917349e55656 : SUCCESS in 4m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/36e21042a1cb4748866a1985d47c6c4a : SUCCESS in 1h 37m 05s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e990120c5e6c40e0be60e3af69603bb3 : SUCCESS in 1h 52m 32s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/afef767da7884f8dba91a013135b59c5 : SUCCESS in 1h 42m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03ffb02901f147c88b40063c7f90baf2 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":7},{"id":"735ab2d20eb0698ab01fc287fee1fffa6c76ca0b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-09 14:23:31.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e133e5173882510253ab4b4134463c62746bdc5c","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-10 01:35:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"562aa29f9c4fda93be44c2f62fb3be705e09a393","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-10 01:36:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cec1ee420b1294297d7c8ad7a0ea83e5179cdedf","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-13 02:33:27.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"67aa7fc07467cdb48d0af9810c4fed6ebf5f525b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-13 07:38:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4a326f5197aa85c7ef75c3e699db1f17d946c453","author":{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"},"date":"2023-11-13 09:35:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9b28619ee7272bd98504945ac442e11784b751c8","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"},"date":"2023-11-13 09:35:51.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_36245\u003e.","accounts_in_message":[{"_account_id":36245,"name":"Asaad Abbasi","display_name":"Asaad Abbasi","email":"asaad.abbasi@gmail.com","username":"AsaadNA"}],"_revision_number":7},{"id":"903780be188953d64ad79497087044637951ac53","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-13 09:56:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"16785ed872cdac411f6ef5f6c503577fadff4f0f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-14 01:51:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2e9196d0b4c6861efb63d613b1a2b6e75770bfb5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-14 02:12:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fe6aeacfe663570acc34d24aca45f722dee5ab16","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-14 13:39:27.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e07721ca94300594e2673252302827a959868392","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-15 01:51:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"65fc264ff99a1b800f4bcd61669d4de32e9bdb71","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-16 14:14:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d8473fb7a3f88ec125b5d52cb7c88700434aed14","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-17 01:37:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"26e227fad9da0e76b540b76b492a8de33819f35d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-17 02:03:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"50f83113c69beba5ad8c74feab5582d0e0148df1","author":{"_account_id":33551,"name":"Ammad Ali","email":"syedammad83@gmail.com"},"date":"2023-11-27 09:47:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54036cc86f15abbd6ad4777c8a0f0ad3b426880b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-27 14:12:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1f3c28a51fe16002982dfa29da36ebab2f45eb74","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-28 02:07:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"efa589a4b13bd2e712e0249bf0930738be436d30","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-29 03:21:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"76dcad54d563604600ec8eec6e23f5168232f455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-05 02:14:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"72d7bf92606d7dfe96e28259712e38ef73f307a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 02:31:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee9d7cd9be64f3d8cf55195421959e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf087691e0204e1f86be6a429bc0745f : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bcff3d7651f245af8f90a6335d5bfe7d : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1710b755c3e44b2e9a5c8d76fe50ddc0 : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3c4ec17609d305c152d30d076e06e1732ea0dfa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 04:15:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89087493836b467fbf79d5b8b904c441\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c13aa6e40b43b98777b421e2ed2a97 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c130645f34fd4b26a6a2e48e10aa8e95 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf578bf27a243cab73af2f3d6ca7612 : SUCCESS in 6m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2f2f3ad8d7e4809a54bfcc36c3667d1 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c037b00909488f92c19fe6fa841296 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/550fd8816d354e3a86bde82ae857fe43 : SUCCESS in 5m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8c89dba875a44ea1b0d63c032ec18c4d : SUCCESS in 1h 31m 29s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9a7b9054a05a4d458ff7ade4d18b2708 : SUCCESS in 1h 37m 19s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/776feddc30c1460db6fd60ba09aaaaf2 : SUCCESS in 1h 19m 16s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/146b826782e9465fb54c6f8b6d743aba : SUCCESS in 1h 15m 37s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/26e3c2ec7bc140d0a60ea69cf8538f02 : SUCCESS in 1h 29m 33s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5b3a7711da344cdea0e3266ab739032f : SUCCESS in 1h 13m 34s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7f86236508e34e1f93b7bb4ef4dcbeab : SUCCESS in 1h 11m 40s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d4a5368708b14f0c8ea7c6b25ed4a151 : SUCCESS in 1h 32m 17s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6983a0c484ea47ce9bb1a537e65171cc : SUCCESS in 1h 15m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/734e6a3d3d3848e8b10200c0213e3823 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":8},{"id":"1efb28339f0527030efbccb96e2d92e747e986a2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-05 06:56:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"503a80bf985abbc3b60cbb7b4905200f5e886b4a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 03:31:00.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4939d110b73098400697cecb8af6703304310976","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-06 03:53:05.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0121a1a39c68505d4b260fcf9df29c797e290b80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 04:06:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c631e8c19d94cf29d380aec6125a04f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/278275a7272f44a4a6035fa4bd4f99b3 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18b3e5722d8a47f981e1920ae9acf626 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75af3f8ceea845a087e2d5a252f0f68f : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"74592fe5af036c7d7de36f0d8e27184c6c2a1d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 06:03:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/490b8700918947c5b19f182ac0c03275\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f19275cb2965474680f33fd8828fdf6e : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6124374746104bef8bcbfc67ac5c1ab0 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b14d6c5974a2422e8d83be2b1ca85493 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bba88139e9334bce94e81bde582fb8f2 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b02562f0e4498a8993bbe1d014a6fd : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02f60367e0b94e6cacb0af765e0a5c37 : SUCCESS in 5m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/edac1d9c33084959a111b3bca050cbbc : SUCCESS in 1h 55m 33s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a232fba2f6da4f56b49455bb8b042fec : SUCCESS in 1h 50m 28s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/178d508599864dfab4329a4520afff37 : SUCCESS in 1h 46m 06s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/94411a3ce6ba42ffa573b5b97274833e : SUCCESS in 1h 42m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e67314fb89dd45e290569bc9b67ae369 : SUCCESS in 1h 37m 01s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/315b46e11e744301b5834bb8374bbe50 : SUCCESS in 1h 43m 46s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/665767716de6452fb79ff6a2bfaa62d5 : SUCCESS in 1h 34m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/38353428b2674a6cac9a1bc7b3d00e7a : SUCCESS in 2h 08m 13s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/740b50c800ec42afbbd1656077c95d1b : SUCCESS in 1h 26m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/938d80e05a4b4b42b451a966ea1ef62c : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":9},{"id":"3baef77a33824a4748e175b489b16363c6bc2aec","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 07:19:20.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7700d05c4714a8008e65de5c29e9ebafd2ab161d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-06 08:03:21.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"0d7bd631ddca708cddc36b443dd39e0cbe68e899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:17:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10af04f48074c8fa3f994ee270ada17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a27d57990a204353b3424f9e11952aee : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1393bf552a0147eebe11d21e218fe734 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63ff6e65ebf64e54bc8e98915dc784c6 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a29d3d53924cc80081a8cce453c12d1cd5dbd8eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 10:37:16.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9de2ff785fd54952acf7a6d4de77e28c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9229dbffa5d41208b99a613e94fe03e : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/368947ff82d84e16bbbac6df3644217c : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/689b3e6a9dc94d5f96fab21e899ff593 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e95826fa28d43cbb3bcf13914098279 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b86144b37447daae16c0f6d8cbeebd : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ff1575451d84c0d98b5ac9ed541c438 : SUCCESS in 5m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/19da92c4ec834855990e1175d34fe2c5 : FAILURE in 2h 31m 11s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/02d842bc66ed4dcbbf9233c775cd38c1 : SUCCESS in 1h 35m 17s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4c46f072a1354873aa3697f146fb7827 : SUCCESS in 1h 45m 05s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3b96694ed7334b32ab8e7dae79a5adcd : SUCCESS in 1h 30m 15s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c5bfbc944724444184321c91495afc52 : SUCCESS in 2h 18m 24s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/266aad7b9d964fcda68bdd405c4d6cbe : SUCCESS in 1h 12m 30s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a018c7ed932c4d0fb6dc98148a8b2f65 : SUCCESS in 1h 09m 46s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/bce427029c694515bc25224506fcb5b3 : SUCCESS in 1h 24m 38s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/450237fcfac943d8af3275f9675e47a3 : SUCCESS in 1h 54m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/faac9bc4b8544bbba7e9750b69ac65ec : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":10},{"id":"3e1e645bdc95b7e30417a75d2746082da079bfc3","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-07 01:40:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4fa3fc94e11c6230c31f8fc0531b65258a9b7d8e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 03:07:20.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"aeddc4b45fa02134df5a83b7b854e65687f03760","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 03:33:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728d6325c8c04b12b9080adb6204a5fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/226d913df1f24a08aee84fdb4b105a7f : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f426e47e8474993a6c98c872b6db19b : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5374c5f9dd3d40e595bd17a2804716f4 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2d50641e8a85f91230986b5176c1ec20348ba891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 04:50:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4addec4f6b984a988e00161aa4797125\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b42b9e164c24284bbe5b2b68a95d4a4 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241cfe1422c648c39fcdc9515c356025 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9113e62fc0e24f6fa9c1f84ee9feba10 : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcf9e48d0be547bab7b241ccded9d179 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8095b95ebb63420aac3215130a906052 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4056f05674c4485ba3ad6ea2bf2a129 : SUCCESS in 5m 29s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/cf0fc033686a4e03be49f24dffc8820b : SUCCESS in 1h 22m 26s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/dff33eb458d2421ea55f96de58a4c49d : SUCCESS in 1h 34m 04s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c67d2f210f43443b975cb529fca082d1 : SUCCESS in 1h 31m 24s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/df7d2103bf344b81b720bc6db789a8ad : SUCCESS in 1h 08m 39s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/92a40d3084fd40b180e5cdadf230414b : SUCCESS in 1h 39m 54s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/05b943fbcae44655af753b6889c3b5ed : SUCCESS in 1h 24m 19s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e0f94d0eae9c4512be646b8c02b8bf92 : SUCCESS in 1h 10m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1ef3915fa9434739b2a0a261aab4c459 : SUCCESS in 1h 38m 55s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/33a8f6ea811e447c9d0707e4e6f82ad7 : SUCCESS in 1h 25m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73c0a7c487e14549acc95bd1fae59a11 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":10},{"id":"d0ccc22e5144f1e8e46122922dc66eda34b6f846","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 05:51:37.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"de6999d66f9512aca2f7395ad64fb59f3715bea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 05:52:06.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"95a4503a6b16c89e14142a5374943e79dbca916b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 07:42:32.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca8844ea4fc24c47a1a775dc71524e96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d42c2fb847343519144f3a1e19970b5 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/285520ccc7384551910bac9a2d5c8e67 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d7ac9222a344de986f4301252d58467 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c799a840f1545eb8e8a5dd1595b4d4c : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43677aa623b74ac2a3ac98385a09b6ea : SUCCESS in 4m 42s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/02789ff5ee4746e5a9d55745afdd5915 : FAILURE in 1h 49m 57s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0635f22d6c164757a315c3f7e172a9b6 : SUCCESS in 1h 35m 49s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7e8c294ff144456b860fe842a357b724 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":10},{"id":"16620a088cc84ccba6507728a61881bd1e5844ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 08:33:40.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_28691\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"_revision_number":11},{"id":"0b2de86a4b9fe7f38fbd89c42a8fde074a6fd990","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 08:45:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd0106f2098452ea374171e0904afcb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96ccee7f3f9c46178e9d3a51697e3ca0 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ea9238cf25c47eb85c769a3eff4f3ca : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9aca17aa6f9c4df8a7355c1f5fcad438 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a5353344b01d71ba09c02bbafaa19da395e69080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 10:56:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39a21824b5ce4981bd2af16cfbbfcd73\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62decf30e2b74b7588d89d22dffa4c5b : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b27d35fa691047df81608e3f8aa4a29f : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2813c61ddf5b438fb22b02d5b7a88e5b : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c73a24698294206a51e549702226c93 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672a49f708d848c5875485aadf720b29 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b477cfdfa084b749c6e20235516b469 : SUCCESS in 4m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9d6ed90ebdc84064b9d25213c616ad36 : SUCCESS in 1h 20m 05s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/00bbe5fe23ac477395c6945ec0eb89d3 : SUCCESS in 2h 15m 43s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/892c6ffc176149af8ceea73bb8feffd7 : SUCCESS in 1h 35m 15s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a8775427ad6b4ac5bbe1deef44ca2a54 : SUCCESS in 1h 06m 27s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/54d44bc18c5841c5ba40bbc015c299e7 : SUCCESS in 1h 34m 18s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f0a3c47962fc4d5fb7ef01256b53fe1c : SUCCESS in 1h 14m 38s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/568932eccaa34ab5a7061e475a970adf : FAILURE in 1h 54m 53s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/48cfcfcd6f4c40ec8a1dbad9279e07a5 : SUCCESS in 1h 24m 00s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/68cc41843cd34bd396264e9425ab1ab8 : SUCCESS in 1h 15m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a03f135b85a4c1c91c4b694128a6e94 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":11},{"id":"92d567e51aea4ca60930b908e9ad2e3d2fea37c9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 02:09:13.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"efcc9fd01fd28d1d9fc570cc0c953a86053f0d0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 01:39:14.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_28691\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"_revision_number":12},{"id":"21ebabdf1d4eac7a3b58814ec9bb4dac8b18c245","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 01:57:25.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce576f9f0d5044538ee63d5bacab68fa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ef0bed0e9fd45b2af8641fd7883a478 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a78a37525c9f4feba91a535ee41cd678 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2ba3e88a0894c76969226230ad07a0b : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"083b5ca011938b82d4e9a8325b423bc7c5666404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 03:58:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98f1a4ea0d604797b031f9d7e7ff0b93\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c2e869416c04454ad8bbb31fa916502 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e99ef79b36f4f9db57a1ad01d20bcb3 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/875d30734b29415abc3db234723a35e2 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d476b0a079ee4d788b918fe30499b73d : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7acfdc213eb4dadb0214651c06a665b : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09f066dd7e774703b3010efead20bea4 : SUCCESS in 5m 43s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c544044e8fdd40efbab34680968e78c9 : SUCCESS in 1h 47m 29s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/92a9ddc520da4677abe6b889dd05a8c9 : SUCCESS in 2h 03m 10s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f5676205e2404c30be8f503228e6de88 : FAILURE in 1h 14m 11s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/789a9014d08b429580676cdad95bfb12 : SUCCESS in 1h 33m 26s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/06093546a7f845e5a2d51c5c03e7f8e3 : SUCCESS in 2h 17m 51s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7ae0caf9318142159511d1f2bf9cff42 : SUCCESS in 1h 36m 35s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/23abc609d3cd4207b544fa55ec712b67 : SUCCESS in 1h 42m 43s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ae4180df3a384c6b9975f8a0e854adbb : FAILURE in 1h 36m 12s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/59c7847419b04e2a9334e08c8e9f4ca9 : SUCCESS in 1h 17m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ecb39c4a64be498b98136dc3461c8474 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":12},{"id":"30fe4aaaf98c505d10e949c0a4db6e14bfd89934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-13 09:12:18.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_28691\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"_revision_number":13},{"id":"e7271f78d73aa81970ba8cbed9461de25e61ec3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 09:19:02.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea40a793881d4cb9892f2f33d9729e0f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcd23fdc01ba4324885f85eecd61871f : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7237da087814449cb0179b953494cf80 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb30a0b76f8d4eaa9f6bd06bab6938e1 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"01ae12cd36f83cbc0bd70415cf27511096a3c9cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 11:02:19.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6ca652e7a024ea3b5af60f329488eba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08503adca63946c180bb1e2050ec4b04 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74fcf75840c4f81ae13f7e651e05798 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbf2d971acab4e029b0d6cf689e4771d : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e33f0af8dfd24b5fa3d22af22dbf987f : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f271b41849664033821e3cda7c738bc6 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1ee321c15ea41f78c1f8b086203bc76 : SUCCESS in 5m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c274ba389d904a95948b70ca1131ea7b : SUCCESS in 51m 34s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e468026fc1c64d3393f957ca2e6fad60 : SUCCESS in 1h 48m 50s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/85ab6131aa154166a250945579f9dbe0 : SUCCESS in 1h 44m 04s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4457b5074e8c4c678dedb3f4f12a2608 : SUCCESS in 48m 50s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8a6c4c5d17e4d97bf98f704f46211ed : FAILURE in 39m 05s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/754dd00f81ad4daf8471fd43888afde7 : SUCCESS in 1h 31m 36s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/0fc5416861574b6fa7829138cd3e6e61 : SUCCESS in 48m 54s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9108a548b51d4fc9a816adaf4141b17a : SUCCESS in 1h 35m 04s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/63975bc2d91846f9b6f0a67c163459bb : SUCCESS in 1h 18m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22f4ac4795504459986d32873456aff7 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":13},{"id":"cecf5ce2320b8db1b73ac07d8ed178ec7e7fa295","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-14 01:20:59.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c74deca78e84423a096f4baf00629f50a878ebc2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 01:29:02.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"907a7902a36a48d340b7978e9c43240ac7ff25c6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 07:24:01.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c80857ad4923e4cc178efe0761c622b4191f5fa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 07:30:56.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d675d0926047df84a721df48406193\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/683ead7700e2440da1ca92471fbfb283 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96c2c1844fb84892a123080328af802f : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60a8a87e1ba4430f93879bc04d76a07d : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8b32a65ebe6fedc198a60a5029c638ac725eeb40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 09:09:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1903f5f53e4937a6961677f0fd2f75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/369f6cd671604033b03b0152611c9b53 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a15f421e714df5af09ee62389e8edd : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c8ad5b84a0b495fbec9927275ff6853 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1dd1e2a38559413b9b72ed0f4e2ba2df : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef078c5299d34bfe83bf950e07dda46c : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eaee90088ef4dd695c8a3d5173fd4dc : SUCCESS in 5m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2e249e8bc0bb4a56b8af25549463a6f2 : SUCCESS in 53m 41s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/90099d3116b6435585d5179667c1804f : SUCCESS in 1h 44m 23s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8112e1e0752b40cc9bcb31ea434af286 : SUCCESS in 1h 25m 42s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8176e9d978b2462189caa49cd976f889 : SUCCESS in 48m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/30b057f06a0f4fe58aadca0b9691c576 : SUCCESS in 1h 31m 40s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b824e09a77ee4ab784cb57ec4d050073 : SUCCESS in 1h 31m 16s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6a491cabe90242faab02da5fe88df320 : SUCCESS in 50m 31s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/65c6d444c7d34e3590f4e34fa37358da : SUCCESS in 1h 34m 07s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/08be345f29b94db8ad6f4eb8bb1b2be8 : SUCCESS in 1h 15m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7fe3c5ad2ecc4e7fbc4b4c10a5b9730a : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":13},{"id":"2ef702cda8b4f97b5531e26aa14539e8dde0f4ec","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 10:25:42.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"6c3c06a0a1ddb05aaec4678b3d9b83fafdf68144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 10:26:21.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"5a8ae9cc279e0dd25a2acffe5aed0ab152a36f68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:58:13.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dc818b8b57745a9b4177ab417c97a48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1417ee13b1b44d79113993156f97463 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dff82249aa14e648500afaf7e3b0d1b : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8824ca2ba90844cb8813e54437c30cd6 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b928d7db30449fada20eca3ce826e0 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2912bc96861c4f84b6d67f66bc6d2a22 : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6abc08bbecad4575b600226fc046401f : FAILURE in 1h 00m 01s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9cae82d50d1c43839c8ad8d57a521e6c : SUCCESS in 54m 58s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7685f50b7d6e49bda9b2293e7afe69af : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":13},{"id":"ee0278fa58c501a4cc615a8e4768eadae5a5fc92","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-15 03:05:31.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"926cb2ca045c19eb6c7be13ec7e7bfe47caa2ca0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 03:13:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859ce9e85253444f89c4a1357450d201\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f875bf6057c40609d512d341d470161 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a18902f7a5964f15911c15592249397e : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c00bae028f4410488952ea426590b91 : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0889ea34bca8816f99e0a32b2c64f6e0da63708e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 04:52:17.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12d7472d1790411a840cd2be8a15e764\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5126661e30cb4018bc6d8793b09d4651 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c69b90bab62e4301ac4fef584be89af7 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a3d090696a4945b9099d9f0353d0cf : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff73f311ea1842fbaad62144a688eea1 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a787b0ab1de49d2af0c9508b79f9920 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeb5049ab5ef4e129970de7440cda23a : SUCCESS in 7m 02s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/304a12fb2da74759a541ac13ffceb7be : SUCCESS in 53m 53s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/226c44bede18418e909ff156ffa9be57 : SUCCESS in 1h 45m 19s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d1a07d57c6f140bba1162237722bb38c : SUCCESS in 1h 28m 22s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c21f25e97881436fa33c0f1ccc77a0aa : SUCCESS in 53m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8c6a048440c34b6686d29878392573a4 : SUCCESS in 1h 39m 45s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/246317abb37c4d539f8884c1d8ac266e : SUCCESS in 1h 15m 45s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ac9969c8d6744829b229857f50a9379d : SUCCESS in 47m 23s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b01c70b8298542eea0497bf743cba658 : FAILURE in 20m 51s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/05406207296d48e184dcbe01b2f1e2d4 : SUCCESS in 1h 11m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79d82bbbe7a04581a6c0daab0df04ad3 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":13},{"id":"a168165fdc3fb8b2d6f4bb5a4d3f9fa380d30e52","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-15 04:54:43.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4428545f9c3905e4e4e938b78335c41b9d440a15","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-15 05:38:47.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"88d9b1bad35780123cf2bd86fd7a9d4e0f3b6b13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 05:45:16.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175d353bdac6473dbab5d7d2f2245326\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d76440e44c704a3c85e93b7366b4479b : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e427b65ae3b1420ca24b020086c867c3 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d29ec9e53e84ae08a472061ec807f73 : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cb7661126f0b1c502f65a57040478dbe23d2f47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 07:36:41.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf942ef6b6ab4a399c315710e127f334\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be9d93499d324519bfa5675a6c1182f4 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83ee9d2961f4b559abf00776e0e36bc : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6a8fbd6780d4a7bab2a61d42e06b12f : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f9e8be6f3b04e91bbcb5ed34a32d437 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf3a49e5a7a41b3a5c184ee992c6741 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57b3080eec3748de9a728fceb32c48e6 : SUCCESS in 4m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0509e681f6164824abbc1c169772c9d5 : SUCCESS in 47m 55s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/aec8bc5170774228bb41f63bca8b173f : SUCCESS in 1h 55m 57s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e51edb51c53c4157bc183a4766637d3e : SUCCESS in 1h 30m 20s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/adf974f7d6b24b4f87b18c4bcf6828c5 : SUCCESS in 46m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/599d12b8ced64d3ba9d427a6c36ae59a : SUCCESS in 1h 44m 05s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7a2b12bdf0b04b24989615c30f690fed : SUCCESS in 1h 16m 43s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/79f7dc2c62d241d8a50e0a49ea55a766 : SUCCESS in 57m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/da82836228e8461d8feb5e227fc39116 : SUCCESS in 1h 26m 30s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3530fcfdec5144f48ae73294dadfcce6 : SUCCESS in 1h 21m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e1bf33d61e1420983d809edf13fb008 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":13},{"id":"7546e28a3ca721ac98a849e174b40ddd994f0666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 07:37:10.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"588a31d5853d713c3f14904cf45d590fd1a624ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 08:20:43.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bcbc2c8253c4cd8af36116328db9420\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3120aa90a3449bcba8f80f7a01321d9 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de3704c4cc24438bbf7defe64d3df7f : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57ef8dcb2ea34b4e873ae0b68bfa727e : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f23319ba784d738df6d92ec8f2583c : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7452e1e45954ec3946476b4e4f64fb5 : SUCCESS in 5m 31s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8a4cab2385c7405f8901b105c4130217 : SUCCESS in 42m 54s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/64b4b9ff7b9344c38c354e21f30d0dc4 : SUCCESS in 39m 58s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/9f60b51bdd004cb697d875b24895ba01 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":13},{"id":"7a87f117ef3052540f0c1600ac665099d635d1a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 08:20:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"86da08f38eb2bfa84f4ede8aa50298148123b358","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 08:21:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b084eca611441fae3148d70131e0bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d4ca529bfb4f95878feb6a5c1500ec : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50438163e54c4cdfbdd6b24382706bf5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"8b6ff821a13888336c141f9b006e648bf75a1744","revisions":{"7a76f3a869715ad5e45f243b6d6e14379aec91ab":{"kind":"REWORK","_number":1,"created":"2021-12-29 16:01:57.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/1"}}},"commit":{"parents":[{"commit":"be86b79119d16ee77f596172f43b0c97cb2617bd","subject":"[Doc] Describe the usage of root_on_create option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be86b79119d16ee77f596172f43b0c97cb2617bd"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a76f3a869715ad5e45f243b6d6e14379aec91ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a76f3a869715ad5e45f243b6d6e14379aec91ab"}]},"branch":"refs/heads/master"},"c4b323e5d55bdbc39e98dbc115ade30e8b5a65e2":{"kind":"REWORK","_number":2,"created":"2023-08-02 08:12:53.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/2"}}},"commit":{"parents":[{"commit":"40363f870d2f8c8a594ef342e5f72e4d1b95e320","subject":"[CI] optimize the image build script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40363f870d2f8c8a594ef342e5f72e4d1b95e320"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-08-02 08:12:37.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4b323e5d55bdbc39e98dbc115ade30e8b5a65e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4b323e5d55bdbc39e98dbc115ade30e8b5a65e2"}]},"branch":"refs/heads/master"},"6d49c5b5afa1e1297bdc7ba2d74e81bcfc168619":{"kind":"REWORK","_number":3,"created":"2023-08-03 02:00:42.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/3"}}},"commit":{"parents":[{"commit":"40363f870d2f8c8a594ef342e5f72e4d1b95e320","subject":"[CI] optimize the image build script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40363f870d2f8c8a594ef342e5f72e4d1b95e320"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-08-03 02:00:11.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d49c5b5afa1e1297bdc7ba2d74e81bcfc168619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d49c5b5afa1e1297bdc7ba2d74e81bcfc168619"}]},"branch":"refs/heads/master"},"edf5079fc0533a2baf3557009b7f0fbf8603b674":{"kind":"REWORK","_number":4,"created":"2023-08-03 06:43:23.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/4"}}},"commit":{"parents":[{"commit":"40363f870d2f8c8a594ef342e5f72e4d1b95e320","subject":"[CI] optimize the image build script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40363f870d2f8c8a594ef342e5f72e4d1b95e320"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-08-03 06:43:07.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edf5079fc0533a2baf3557009b7f0fbf8603b674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edf5079fc0533a2baf3557009b7f0fbf8603b674"}]},"branch":"refs/heads/master"},"a2e21bdae65531b61ed29891f37e1357d05a0559":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-03 04:26:14.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/5"}}},"commit":{"parents":[{"commit":"e998b6886602575127ebe613e56cee3a5a01c6c6","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-11-03 04:24:37.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2e21bdae65531b61ed29891f37e1357d05a0559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2e21bdae65531b61ed29891f37e1357d05a0559"}]},"branch":"refs/heads/master"},"5dc71f41e436c6912050bd0cbe5c1ecc63fa7344":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-03 07:06:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/64/823164/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/6"}}},"commit":{"parents":[{"commit":"a60b63d7abd444c2ef44bfde6369d826913e1e1b","subject":"Merge \"[CI] Improve CI tests for mysql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a60b63d7abd444c2ef44bfde6369d826913e1e1b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-11-03 07:06:57.000000000","tz":0},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5dc71f41e436c6912050bd0cbe5c1ecc63fa7344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5dc71f41e436c6912050bd0cbe5c1ecc63fa7344"}]},"branch":"refs/heads/master","description":"Rebase"},"6ad9c14483dd0f2079d1d65860afbe05fef8c5d4":{"kind":"REWORK","_number":7,"created":"2023-11-08 04:28:48.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/7"}}},"commit":{"parents":[{"commit":"e998b6886602575127ebe613e56cee3a5a01c6c6","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-11-08 04:28:35.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ad9c14483dd0f2079d1d65860afbe05fef8c5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ad9c14483dd0f2079d1d65860afbe05fef8c5d4"}]},"branch":"refs/heads/master"},"d6aa04c0d568c207a678b5f8de077d7706789441":{"kind":"REWORK","_number":8,"created":"2023-12-05 02:14:43.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/8"}}},"commit":{"parents":[{"commit":"e998b6886602575127ebe613e56cee3a5a01c6c6","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-12-05 02:14:28.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6aa04c0d568c207a678b5f8de077d7706789441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6aa04c0d568c207a678b5f8de077d7706789441"}]},"branch":"refs/heads/master"},"321fc5a720d31b1e373a1c8aa2e3184e4bdb53cb":{"kind":"REWORK","_number":9,"created":"2023-12-06 03:53:05.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/9"}}},"commit":{"parents":[{"commit":"e998b6886602575127ebe613e56cee3a5a01c6c6","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-12-06 03:52:54.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/321fc5a720d31b1e373a1c8aa2e3184e4bdb53cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/321fc5a720d31b1e373a1c8aa2e3184e4bdb53cb"}]},"branch":"refs/heads/master"},"8a5b837a47a6496b3ce43a88449f6d6cd95acf0e":{"kind":"REWORK","_number":10,"created":"2023-12-06 08:03:21.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/823164/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/10"}}},"commit":{"parents":[{"commit":"e998b6886602575127ebe613e56cee3a5a01c6c6","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-12-06 08:03:09.000000000","tz":420},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a5b837a47a6496b3ce43a88449f6d6cd95acf0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a5b837a47a6496b3ce43a88449f6d6cd95acf0e"}]},"branch":"refs/heads/master"},"a92836d0aae6b691499ac37a737305967ad18d71":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-12-07 08:33:40.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/64/823164/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/11"}}},"commit":{"parents":[{"commit":"aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9","subject":"Merge \"Remove unnecessary ceilometer service overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-12-07 08:33:40.000000000","tz":0},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a92836d0aae6b691499ac37a737305967ad18d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a92836d0aae6b691499ac37a737305967ad18d71"}]},"branch":"refs/heads/master","description":"Rebase"},"afbd1fab2019d2f91379481327e4674de511ab7c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-12-12 01:39:14.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/64/823164/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/12"}}},"commit":{"parents":[{"commit":"21b4cc2cf7ca3a09bdefb3654dcbb07043ea2ee6","subject":"Don\u0027t configure ip route for slaac ipv6 subnet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21b4cc2cf7ca3a09bdefb3654dcbb07043ea2ee6"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-12-12 01:39:14.000000000","tz":0},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afbd1fab2019d2f91379481327e4674de511ab7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afbd1fab2019d2f91379481327e4674de511ab7c"}]},"branch":"refs/heads/master","description":"Rebase"},"8b6ff821a13888336c141f9b006e648bf75a1744":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-12-13 09:12:18.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/64/823164/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/823164/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/823164/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/823164/13"}}},"commit":{"parents":[{"commit":"22469809ecca7570953f4d9d67529316939cc451","subject":"Merge \"[CI] Fix Ci failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/22469809ecca7570953f4d9d67529316939cc451"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-12-29 15:59:13.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-12-13 09:12:18.000000000","tz":0},"subject":"Cinder Support For Boot Volume","message":"Cinder Support For Boot Volume\n\nWhen use cinder to store data dir of database, also create rootdisk in cinder.\n\nStory: #2009245\nTask: #43418\nChange-Id: Ia5841222c7a70cb0c88078575b4d8b4f7988d5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b6ff821a13888336c141f9b006e648bf75a1744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b6ff821a13888336c141f9b006e648bf75a1744"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
