)]}'
{"id":"openstack%2Ftrove~823632","triplet_id":"openstack%2Ftrove~master~I34d47e709a9a7478cb6a85d6e6e37da3f92e40b0","project":"openstack/trove","branch":"master","topic":"inject_userdata_config","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2022-02-13 11:25:23.000000000","reason":"Change was submitted"},"33822":{"account":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"last_update":"2022-02-13 10:42:45.000000000","reason":"Qiannan Lyu replied on the change"}},"hashtags":[],"change_id":"I34d47e709a9a7478cb6a85d6e6e37da3f92e40b0","subject":"Adapt to file injection deprecation in nova","status":"MERGED","created":"2022-01-06 07:27:42.000000000","updated":"2022-02-13 11:26:42.000000000","submitted":"2022-02-13 11:25:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823632-inject_userdata_config","meta_rev_id":"07e7e46543b552b7eff3e00fb7f916e32abe29fc","_number":823632,"virtual_id_number":823632,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-13 11:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-13 10:59:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":1,"date":"2022-02-13 10:42:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-13 10:59:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 08:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 02:58:47.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"CC"},{"updated":"2022-02-03 02:58:47.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"state":"REVIEWER"},{"updated":"2022-02-13 10:59:37.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"43e031e9a279e0a2e6e81c2c1dca87337cb94b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-01-06 07:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b93c108c790a23339b9b88d038a2e438a5c53a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 08:12:44.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb874439b5224d0ab765a3d83f028c98 : FAILURE in 7m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185100b4272f432b9a04d9eae033eb2f : FAILURE in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b35ec77d33894236b70aacb73f9e7412 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf95256ee659444f813917b31cf81392 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380e126fe4734bc094f03cf1477ece73 : SUCCESS in 6m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a15c2bab96c448c858e0478c450d3fc : SUCCESS in 6m 37s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e50b049d15a54223985c1b1a35c292c1 : FAILURE in 19m 05s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/68988aa5bae1480e9e0e582c66f4db6a : FAILURE in 19m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c82a77de26546cda91a23e1f8a5efa8 : FAILURE in 37m 43s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c7281b329e1042f7a8a5245acb1a9402 : FAILURE in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9b6a4bdf27393402196588021efbc8fc426a569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-01-06 09:15:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c45ab7c41e205fda76ebcdc1a745a74a1017b22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 09:57:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53d0272de3784287923894d3c7514ed4 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57493307af494de68bf5d4449f24c534 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11b41215bc2844d5a58dbf57e74b44d3 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e084a9a3095a4777a45eea1458fb77e9 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ff935969394f129b485f32a48239aa : SUCCESS in 5m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d518787fc984726890dced41063c025 : SUCCESS in 6m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/154ede7264544aada146bdf9744f30b4 : FAILURE in 25m 21s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e603966885e34a4d9dd80f3dd18a14b7 : FAILURE in 18m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5a65d02945444d5a8fabdfb31f49085 : FAILURE in 33m 26s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/08f67dc0c4624e57aed6a6e1c21e1554 : FAILURE in 18m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"341ece5c18cef343ae0f292ae95f61b87f8e1c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-01-24 10:06:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"460bf95dc7e5695cafd7daa50de09f7dd208384c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 10:32:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a4276955af418cae3c6cbda6d043de : FAILURE in 7m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32cae3b5dc0441568715c7cf27e916bc : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/643a0763b840475fb46e4a7fae3ca2d7 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fc97b8d5b2143a8b31b8a990d5fe0f1 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0684560e7b15480eb02922e2e6bb6e89 : SUCCESS in 7m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c85e43c3ad0c476bb12228f59ad7f0e4 : SUCCESS in 6m 12s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/83974bca716b4ba3a79b18aac00a9bf2 : FAILURE in 17m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/34a65dbe3a0a439ba2589376e8008c9d : FAILURE in 18m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/016a2347feec43ea945bb17fa9c395f2 : FAILURE in 24m 15s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ff4bd8cbbdf8414790d34470c4c931dd : FAILURE in 17m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a41d67535f87da00c38e9a44d68045c9b099667","author":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"date":"2022-02-13 10:42:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3f56219469aad45ff4ea5ddd7a454e3f3fd6d527","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2022-02-13 10:59:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"699161f41d3fa065575218ffbdd1e126e44861b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:00:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8914273119805ab2cbc2e0c62323111c965d8577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:25:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e877bcf56a499fa2db9b1289bff2b6 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c81a48bcff47bd8a9bbd1770b614bb : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/226943052fd64ac39358684cb3be5677 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000830af127a4550a138f8c1e63410ca : SUCCESS in 5m 12s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2c311698074c407c851409c9b8767e57 : FAILURE in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"536890fec780ea1fbfe344466f803fd7f02bc77c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:25:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"07e7e46543b552b7eff3e00fb7f916e32abe29fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:26:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1c54efaff14c78af2b773d01777e8b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"99df090a857f48ae27e9c361f6835fc2297a0498","revisions":{"1318445b3962dc7b149fd10c41ea61552512b08e":{"kind":"REWORK","_number":1,"created":"2022-01-06 07:27:42.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/32/823632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/823632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/823632/1"}}},"commit":{"parents":[{"commit":"be86b79119d16ee77f596172f43b0c97cb2617bd","subject":"[Doc] Describe the usage of root_on_create option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be86b79119d16ee77f596172f43b0c97cb2617bd"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-06 07:27:25.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-06 07:27:25.000000000","tz":420},"subject":"Adapt to file injection deprecation in nova","message":"Adapt to file injection deprecation in nova\n\nStory: #2009770\nTask: #44247\nChange-Id: I34d47e709a9a7478cb6a85d6e6e37da3f92e40b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1318445b3962dc7b149fd10c41ea61552512b08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1318445b3962dc7b149fd10c41ea61552512b08e"}]},"branch":"refs/heads/master"},"02c1efc47af7d405c1d019114e22d5cac9123d66":{"kind":"REWORK","_number":2,"created":"2022-01-06 09:15:36.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/32/823632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/823632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/823632/2"}}},"commit":{"parents":[{"commit":"be86b79119d16ee77f596172f43b0c97cb2617bd","subject":"[Doc] Describe the usage of root_on_create option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be86b79119d16ee77f596172f43b0c97cb2617bd"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-06 07:27:25.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-06 09:15:23.000000000","tz":420},"subject":"Adapt to file injection deprecation in nova","message":"Adapt to file injection deprecation in nova\n\nStory: #2009770\nTask: #44247\nChange-Id: I34d47e709a9a7478cb6a85d6e6e37da3f92e40b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02c1efc47af7d405c1d019114e22d5cac9123d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02c1efc47af7d405c1d019114e22d5cac9123d66"}]},"branch":"refs/heads/master"},"99df090a857f48ae27e9c361f6835fc2297a0498":{"kind":"REWORK","_number":3,"created":"2022-01-24 10:06:23.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/32/823632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/823632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/823632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/823632/3"}}},"commit":{"parents":[{"commit":"be86b79119d16ee77f596172f43b0c97cb2617bd","subject":"[Doc] Describe the usage of root_on_create option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be86b79119d16ee77f596172f43b0c97cb2617bd"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-06 07:27:25.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2022-01-24 10:06:05.000000000","tz":420},"subject":"Adapt to file injection deprecation in nova","message":"Adapt to file injection deprecation in nova\n\nStory: #2009770\nTask: #44247\nChange-Id: I34d47e709a9a7478cb6a85d6e6e37da3f92e40b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99df090a857f48ae27e9c361f6835fc2297a0498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99df090a857f48ae27e9c361f6835fc2297a0498"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
