)]}'
{"id":"openstack%2Ftrove~832978","triplet_id":"openstack%2Ftrove~master~I9a0c555f94f97c5d6bfee7e9dde0c96930b2c27b","project":"openstack/trove","branch":"master","topic":"add-zed-python-jobtemplates","hashtags":[],"change_id":"I9a0c555f94f97c5d6bfee7e9dde0c96930b2c27b","subject":"Add Python3 zed unit tests","status":"MERGED","created":"2022-03-10 08:20:53.000000000","updated":"2022-03-19 10:00:45.000000000","submitted":"2022-03-19 09:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832978-add-zed-python-jobtemplates","meta_rev_id":"6b12913b29e83f2c919fcabee769dd87ca1725fe","_number":832978,"virtual_id_number":832978,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-19 09:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-19 05:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-19 05:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 09:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-19 05:04:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"895285ec18f8e29c4c24029dd313a29b3a7ebfde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-10 08:20:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca5473588fe0b06f6f67408f2c0e18eaad8648b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 09:16:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3cb891138ab4bb39b1e7d7d42044970 : FAILURE in 9m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574e13769db8453fb4d1a629202f3ba7 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97856acc5d0b4c16bb5e9015fca68938 : FAILURE in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a295f2e23275437dbfb08c3dc309cad3 : FAILURE in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fd00afc0ee64fe8ac370e6e4d390924 : FAILURE in 15m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd4486f337042ca958334e0548c2b59 : SUCCESS in 9m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/888509e235e8439c8cf370a21b425189 : SUCCESS in 8m 32s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1b5335e6dae049b1aedebce85f365004 : FAILURE in 28m 59s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2dcb7b79e06b412999c13417e8367e96 : FAILURE in 30m 04s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ac2fafbb2384d459cda908d506838c4 : FAILURE in 40m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f0cbefc3f0b344a4888783fc9fb17a04 : FAILURE in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04a9c912fd71db082ae22567fde83a7341f15710","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-03-19 05:04:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26df19bef1d6e3186f84c4fa43a50c9b26592ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 05:41:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ee0e89a79ef44bc9ed77665251acc8d : FAILURE in 7m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b89a1f6db084fa28b998781a5722bfd : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a44f3c6462314098a2a49035c14da9e4 : FAILURE in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/241867edc0e740a9bd0ab95681d3b451 : FAILURE in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9e1ebadb87445fd8672e236a64fc829 : FAILURE in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaee86aa7b0b4237a95dcd4a49b64591 : SUCCESS in 6m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/182e31cb4c8845e096feb6c096323db8 : SUCCESS in 6m 20s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1475a10d81f943f4ad698a6302350aa2 : FAILURE in 25m 13s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b1ddaccd6a174649985427c5784d9a6a : FAILURE in 27m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00be26b250f944969a067ab4f45880fe : FAILURE in 25m 18s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3e6e749dd72e45ba916cda5415ccb3e5 : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13df410bd7f768d3da03151e6627035b951e167a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-03-19 08:24:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10196638ebcf0e6761980946482bb3e2ac5e7895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 09:20:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41be53b31bd145d888f4e7c81006fc8d : FAILURE in 8m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35eb877908b64c818aa88387e38e1e20 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71d3ae02e50c498ea1208c85a61d7a8b : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5483d75f5c29410c84448515666b1b1a : SUCCESS in 8m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/732e1af58ffe44f2a7c80192f0b03349 : FAILURE in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489dd1c1018840f881e3c38d2f8c1036 : SUCCESS in 9m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2d45c5bb521d4d6d95aee22914dbd96f : SUCCESS in 8m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bf4e62b86e6d44238d39cdfc5c2e40bb : FAILURE in 35m 55s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a8695fa56ca54c24a06a2c8f9e20da0a : FAILURE in 35m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7750f324c004248bb5d194122bbb94d : FAILURE in 29m 22s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7d3db84dffb8449c84d85192f6b6ab0b : FAILURE in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d24a1458359141947f4e360e38f03a5c3a9c2a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 09:21:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1197bfb875c5a10355da43452b910b7fd978a993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 09:56:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b03cb00ffa475f918ce7a068ecbdd8 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99a4e71b198349b3ae2c9bda8f58d84f : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64f1cb1623714b7f8f9f5fc167cff9dc : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b4867b50ee4f49bbfab46cbbd2ceb5 : SUCCESS in 7m 09s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3284c1aea66b44b984aa5b8dfade620c : FAILURE in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8efba31789eeb2f539af54808a6d4be179c81317","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 09:56:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6b12913b29e83f2c919fcabee769dd87ca1725fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 10:00:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c6e9d0056b443ab7c493db965f4a92 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9a81fb65768333ce43ab918e6e96aa54b7e7b8a","revisions":{"f9a81fb65768333ce43ab918e6e96aa54b7e7b8a":{"kind":"REWORK","_number":1,"created":"2022-03-10 08:20:53.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/78/832978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/832978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/832978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/832978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/832978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/832978/1"}}},"commit":{"parents":[{"commit":"391bca927d5fdcaaddc41815fd54552a1f1b2cd6","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/391bca927d5fdcaaddc41815fd54552a1f1b2cd6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-10 08:20:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-10 08:20:52.000000000","tz":0},"subject":"Add Python3 zed unit tests","message":"Add Python3 zed unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for zed.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I9a0c555f94f97c5d6bfee7e9dde0c96930b2c27b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9a81fb65768333ce43ab918e6e96aa54b7e7b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9a81fb65768333ce43ab918e6e96aa54b7e7b8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
