)]}'
{"id":"openstack%2Ftrove~833186","triplet_id":"openstack%2Ftrove~master~I2b38e31c14e68aa5ee5bd1130c5147135fef5df1","project":"openstack/trove","branch":"master","topic":"use-project-not-tenant","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2022-03-11 02:12:31.000000000","reason":"Bo Tran replied on the change"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-10-10 12:09:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2b38e31c14e68aa5ee5bd1130c5147135fef5df1","subject":"WIP: use project, not tenant","status":"ABANDONED","created":"2022-03-10 22:19:56.000000000","updated":"2023-10-10 12:09:25.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d744f56e15610c1312a1330d631331a2331238d6","_number":833186,"virtual_id_number":833186,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-10 23:01:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 23:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 02:11:19.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"ec7f2e9cd7064769f2ef8ae7919468c83d3aa0bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-03-10 22:19:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f24477602de9cba0b2ba778d282a2c11ec5081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 23:01:48.000000000","message":"Patch Set 1: Verified-1\n\n(34 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2f07d89307c940aa8a9678e01e66d839 : FAILURE in 8m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18c72eacf0b6447c9cd8401cb8411f5c : FAILURE in 9m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99244b25c0f44e994fee78e066a8a68 : FAILURE in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/559b6214fd414f0ead1b05278d3d5e95 : FAILURE in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b300f7f04324392aa374b77cc1d5ba5 : FAILURE in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0200b7ddbd4cc483392af60f42d286 : SUCCESS in 8m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43281aa411994cfabd43747b524ac9b9 : SUCCESS in 8m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/38727ad3ce10497e8a2f66f791792859 : FAILURE in 38m 11s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7a39eb37990e4339911f1e230489b9c2 : FAILURE in 39m 29s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dae43dd33b1a44b4b52eb035f82e3131 : FAILURE in 34m 04s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/663ef7fb909d47e2a6c5aab6144143a3 : FAILURE in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71bba9f4d4cd3233c8864364d1ec1a103cd92570","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-03-11 02:12:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d744f56e15610c1312a1330d631331a2331238d6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-10-10 12:09:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38a9d4ef83d95e274a40017c366c102382e690fd","revisions":{"38a9d4ef83d95e274a40017c366c102382e690fd":{"kind":"REWORK","_number":1,"created":"2022-03-10 22:19:56.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/86/833186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/833186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/833186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/833186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/833186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/833186/1"}}},"commit":{"parents":[{"commit":"39322f5ad88cf5eef003af74294676a5563a8296","subject":"Merge \"Improve the detect backup was created when parse log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39322f5ad88cf5eef003af74294676a5563a8296"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-03-10 22:14:41.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-03-10 22:19:14.000000000","tz":60},"subject":"WIP: use project, not tenant","message":"WIP: use project, not tenant\n\nThis is a first naive approach, where db migrations aren\u0027t\neven taken into account. This is just to see how Trove would\ndo with a simple search/replace...\n\nChange-Id: I2b38e31c14e68aa5ee5bd1130c5147135fef5df1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38a9d4ef83d95e274a40017c366c102382e690fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38a9d4ef83d95e274a40017c366c102382e690fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
