)]}'
{"id":"openstack%2Ftrove~835492","triplet_id":"openstack%2Ftrove~stable%2Fyoga~I69c7098cc0d61fbbba1dbf2eca87df0dd6fd70ba","project":"openstack/trove","branch":"stable/yoga","topic":"bug/2009906-stable/yoga","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-03-29 04:19:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69c7098cc0d61fbbba1dbf2eca87df0dd6fd70ba","subject":"Removes the deprecated argument tenant from TroveContext","status":"MERGED","created":"2022-03-28 18:18:34.000000000","updated":"2022-03-29 04:20:33.000000000","submitted":"2022-03-29 04:19:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"835492-bug/2009906-stable/yoga","cherry_pick_of_change":832786,"cherry_pick_of_patch_set":3,"meta_rev_id":"a735c68cfe8c1b01a187e2b3633143e449784f33","_number":835492,"virtual_id_number":835492,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-29 04:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-03-29 02:28:48.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-03-29 02:28:48.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}],"CC":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 18:18:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"CC"},{"updated":"2022-03-28 20:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-29 02:28:48.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"0c6369df1e06bd43045623b341aea18f2d83b95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-28 18:18:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c598125d9bf50912ebadd46ae79b2b7a63b1d944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 20:27:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2570594a4bc4c45bbb93d93446a4882 : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de94f7466c7141cf9af217359614a9bd : FAILURE in 8m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ce2f61a92b43d7bf8252b78e8c0ab3 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/704a84432bd0435183f640c9562ccdf3 : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb288f706444e93bfb3ff31517aefec : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f152fbc64d459699f996b3ef156ac1 : SUCCESS in 7m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4b722fab4c74b24bbba7eaf92677779 : SUCCESS in 7m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7f35a1b00cfe4ed18b34ce7da34adcff : FAILURE in 2h 05m 59s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d8706f8552024fe98ec188d74668c244 : FAILURE in 2h 04m 32s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/782ff6ee43a04592879941cdba2666e2 : FAILURE in 40m 51s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3984cb8e7c274f468cb73349b898ccae : FAILURE in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22c234ef893bbef7797bc1b518d4932a233a8376","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-28 20:35:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90125042ffd6db5aceceed3788e517cd012c6583","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-03-29 02:28:48.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2\n\nEmergency approving at the direction of the OpenStack TC and Release Managers: https://lists.openstack.org/pipermail/openstack-discuss/2022-March/027865.html","accounts_in_message":[],"_revision_number":1},{"id":"b9d4de09a3c66a5e8fd2e54d5457c43af647167f","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-03-29 03:31:43.000000000","message":"Patch Set 1:\n\nrecheck in order to attempt to figure out why zuul refuses to enqueue this into the gate pipeline","accounts_in_message":[],"_revision_number":1},{"id":"65d97bf9a63b975a52025e0ac1c7ee01bb78072e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 03:32:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a793b47bb7b8ef6be623a7ac1730b6e86a9b80b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 04:19:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f7c972378a145c9a6e6f9951e5fbde7 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b090bc396f4468facec458288d4285a : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6c4e02cffd4422e96dc4b4a281171a8 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37971eee0b7d4930baf47dcfb507ec42 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900ce483222743eb97c5e7faa363f119 : SUCCESS in 7m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/492e9c6a310b42b08d9c8ea90d78e009 : FAILURE in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53de862a5079760d0ae576c5acc4b337ed93e827","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 04:19:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a735c68cfe8c1b01a187e2b3633143e449784f33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 04:20:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6cf93a55f84d03addf6b5d0fdf6246 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2872e91c04e9f60e1c648964a35f5d875493ef7a","revisions":{"2872e91c04e9f60e1c648964a35f5d875493ef7a":{"kind":"REWORK","_number":1,"created":"2022-03-28 18:18:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/92/835492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/835492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/835492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/835492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/835492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/835492/1"}}},"commit":{"parents":[{"commit":"39322f5ad88cf5eef003af74294676a5563a8296","subject":"Merge \"Improve the detect backup was created when parse log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39322f5ad88cf5eef003af74294676a5563a8296"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-03-09 12:17:11.000000000","tz":540},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-28 18:18:34.000000000","tz":0},"subject":"Removes the deprecated argument tenant from TroveContext","message":"Removes the deprecated argument tenant from TroveContext\n\nThe tenant argument of RequestContext in oslo.context had been\ndeprecated long time ago and it was finally removed in\noslo.context-4.0.0. We should remove the tenant argument of\nTroveContext that derives from RequestContext and we should also\nupdate the requirements.txt in the master branch.\n\nTask: 44723\nStory: 2009906\nChange-Id: I69c7098cc0d61fbbba1dbf2eca87df0dd6fd70ba\n(cherry picked from commit 559d6255e5feb73b90468a25a8ba99650ad50bf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2872e91c04e9f60e1c648964a35f5d875493ef7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2872e91c04e9f60e1c648964a35f5d875493ef7a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
