)]}'
{"id":"openstack%2Ftrove~836866","triplet_id":"openstack%2Ftrove~master~I7ee128afad80b92bdeabb45f719d19a99badd8a7","project":"openstack/trove","branch":"master","topic":"story/2009958","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-04-18 12:54:33.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-04-18 12:22:01.000000000","reason":"wu.chunyang replied on the change"}},"hashtags":[],"change_id":"I7ee128afad80b92bdeabb45f719d19a99badd8a7","subject":"Validates \"network_id\" param creating a cluster","status":"MERGED","created":"2022-04-06 17:06:43.000000000","updated":"2022-04-18 14:22:21.000000000","submitted":"2022-04-18 12:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836866-story/2009958","meta_rev_id":"bf6a5992cef42702d00963cbfab585f4982adf52","_number":836866,"virtual_id_number":836866,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-18 12:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-18 12:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-18 12:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 19:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-09 18:13:22.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"caaa9e6fa63d770942d6733fa82ee348f54146c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-06 17:06:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92a55fcb18f0adb54d209deb4438e7d47fc344ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 19:12:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/883a1b01a48c4385af774aa597d8474a : FAILURE in 8m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de96762968c649bc9a5d614eb413d37d : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7815e28fc0a4ecf85a7699d3d621682 : FAILURE in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8da3f20938ab4630b9e39188628f4a5a : FAILURE in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfdbc820a4f248edba8a181c1f02ae58 : FAILURE in 13m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd4fd15626e4a3b8edae862ddd1e315 : SUCCESS in 6m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce5f03155cff46339484ad2d9800aab4 : SUCCESS in 6m 56s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6902748c1b7e44f8b5cfd1ae07095e72 : FAILURE in 2h 03m 38s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/74870dacc1d044a2a2d41132ed71bd25 : FAILURE in 31m 07s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66370d13bcd64a7b9726eacb28e88379 : FAILURE in 52m 52s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f0ee0c9b6d3a402ba2c70420caa7feb9 : FAILURE in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfe53b8022a360cb9f81348086ed56b64c4c48dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-09 14:52:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81162a0ebf4a01935d9ae4a4d28966709f766252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 17:02:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df73cb25676d482c9e3a49da17fb67a8 : FAILURE in 7m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413559cdac3a4090855199059a8a3c40 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de306389139240bc932ca10970fd6382 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11e749dbacda450e8836e04611ff201f : SUCCESS in 8m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f4573a7bd1045d68a2568f463294ea6 : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2322e93e7dd429a964de381f0e454de : SUCCESS in 6m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a1345e12d0947179ac50b54f36328b5 : SUCCESS in 6m 37s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e5e66c98d03d4b88ae0e245b38b22a01 : FAILURE in 2h 07m 04s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/562901acc547406f9b5b504762a89d26 : FAILURE in 2h 09m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bbb1e8514f747a6a96168f694a5d1e2 : FAILURE in 1h 12m 24s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9717009f91e3475aab4aeab3d631156d : FAILURE in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35b62a6b20e16721355da7599e9cd5a3c0d08080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-09 18:03:12.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fa4cea6f2c06838f2ddf6ab00ce869505796a868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-09 18:08:31.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ca3720a2e0165ef62eecc3ac80bcd4bf1a8be919","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-09 18:13:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"542dd2e21955a42743e637dad5c28ca5b3423a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 18:44:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8756e15978c45baa3f7a64284a5d492 : FAILURE in 6m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5488d8ad8b004dd4a0b91531fbd58b52 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63229eff37164d9aa8c6b029f037013a : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d0845b3e564a898d141f391c80ed7a : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4d0293d98d64666b273cfd689931000 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9131f7f782435385b3bd49e8d32f0a : SUCCESS in 5m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d5ecae89e2047678ae69061c50bc06e : SUCCESS in 5m 00s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3a0aae3d10f94e389075d3a5a819826d : FAILURE in 26m 25s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b2d3b30575394c21ab6ec818d365a435 : FAILURE in 25m 24s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5a496be7afa4360a2998a77fa38ad86 : FAILURE in 25m 40s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7d8bbf03a0804d09a9099e4ecb66220d : FAILURE in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7328840dd5a525aaf6a07f2f86e91f9f521926fd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-18 12:22:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"512cf6a4270fe31ac6d57f623a34872d40d1d27f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:22:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9badf9821c4380ae9d89b645749811c3ab8aee3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:54:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d2f5141106d40cbb832c7f4527aa70e : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b17eb308e2a43f5b38d5b5b2bdbbef7 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a70264ffcdb493a94ce403b1020591c : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef3d48b750e467eb007da762a3416cc : SUCCESS in 5m 31s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ddd09a6b1c1a44698c6fd95f2fb68732 : FAILURE in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c8c9de8e515905d4972b7f4495aee1dba5da445","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:54:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"765c4042e715e500701b517a6e36233f109e83fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:56:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7c76aacd6c4c09aa34326ac028a710 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4},{"id":"bf6a5992cef42702d00963cbfab585f4982adf52","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-18 14:22:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9e518a5513927bd96ea8e8d6c6395282ea626848","revisions":{"4c260e8fbb2bc45109d618b1f42418bb2adc88d5":{"kind":"REWORK","_number":1,"created":"2022-04-06 17:06:43.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/836866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/836866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/836866/1"}}},"commit":{"parents":[{"commit":"92471bd4f39a0805bd1cbd732fa12d530e607c8f","subject":"Adds docker daemon config to injected files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/92471bd4f39a0805bd1cbd732fa12d530e607c8f"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-06 17:05:18.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-06 17:05:49.000000000","tz":540},"subject":"Validates \"network_id\" param creating a cluster","message":"Validates \"network_id\" param creating a cluster\n\nThis PR changes Clusters API validator to validate a request parameter\n\"network_id\" instead of \"net-id\". When creating a cluster, Clusters\nAPI internally calls Instaces API to create a instance. Instances API\nhas replaced the request parameter \"net-id\" with \"network_id\" by this\ncommit[1].\n\n[1]: https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372\n\nTask: 44967\nStory: 2009958\nChange-Id: I7ee128afad80b92bdeabb45f719d19a99badd8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c260e8fbb2bc45109d618b1f42418bb2adc88d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c260e8fbb2bc45109d618b1f42418bb2adc88d5"}]},"branch":"refs/heads/master"},"d4db33dc3deebed9c287fc08526dc3eb253a4f7d":{"kind":"REWORK","_number":2,"created":"2022-04-09 14:52:18.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/836866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/836866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/836866/2"}}},"commit":{"parents":[{"commit":"92471bd4f39a0805bd1cbd732fa12d530e607c8f","subject":"Adds docker daemon config to injected files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/92471bd4f39a0805bd1cbd732fa12d530e607c8f"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-06 17:05:18.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-09 14:51:50.000000000","tz":540},"subject":"Validates \"network_id\" param creating a cluster","message":"Validates \"network_id\" param creating a cluster\n\nThis PR changes Clusters API validator to validate a request parameter\n\"network_id\" instead of \"net-id\". When creating a cluster, Clusters\nAPI internally calls Instaces API to create a instance. Instances API\nhas replaced the request parameter \"net-id\" with \"network_id\" by this\ncommit[1].\n\n[1]: https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372\n\nTask: 44967\nStory: 2009958\nChange-Id: I7ee128afad80b92bdeabb45f719d19a99badd8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d4db33dc3deebed9c287fc08526dc3eb253a4f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d4db33dc3deebed9c287fc08526dc3eb253a4f7d"}]},"branch":"refs/heads/master"},"7f1b469ddfa16ec38be19d8e1fd7104f2c86a9f8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-09 18:03:12.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/836866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/836866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/836866/3"}}},"commit":{"parents":[{"commit":"92471bd4f39a0805bd1cbd732fa12d530e607c8f","subject":"Adds docker daemon config to injected files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/92471bd4f39a0805bd1cbd732fa12d530e607c8f"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-06 17:05:18.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-09 18:03:12.000000000","tz":0},"subject":"Validates \"network_id\" param creating a cluster","message":"Validates \"network_id\" param creating a cluster\n\nThis PR changes Clusters API validator to validate a request parameter\n\"network_id\" instead of \"net-id\". When creating a cluster, Clusters\nAPI internally calls Instaces API to create a instance. Instances API\nhas replaced the request parameter \"net-id\" with \"network_id\" by this\ncommit[1].\n\nThe original problem is instance creation error when creating a\ninstance that belongs to a cluster. Troves\u0027s Instances API has\nchanged a parameter name from \"net-id\" to \"network_id\", but Clusters\n API still uses the \"net-id\" parameter name.\n\n[1]: https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372\n\nTask: 44967\nStory: 2009958\nChange-Id: I7ee128afad80b92bdeabb45f719d19a99badd8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f1b469ddfa16ec38be19d8e1fd7104f2c86a9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f1b469ddfa16ec38be19d8e1fd7104f2c86a9f8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9e518a5513927bd96ea8e8d6c6395282ea626848":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-09 18:08:31.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/836866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/836866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/836866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/836866/4"}}},"commit":{"parents":[{"commit":"92471bd4f39a0805bd1cbd732fa12d530e607c8f","subject":"Adds docker daemon config to injected files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/92471bd4f39a0805bd1cbd732fa12d530e607c8f"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-06 17:05:18.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-09 18:08:31.000000000","tz":0},"subject":"Validates \"network_id\" param creating a cluster","message":"Validates \"network_id\" param creating a cluster\n\nThis PR changes Clusters API validator to validate a request parameter\n\"network_id\" instead of \"net-id\". When creating a cluster, Clusters\nAPI internally calls Instaces API to create a instance. Instances API\nhas replaced the request parameter \"net-id\" with \"network_id\" by this\ncommit[1].\n\nThe original problem is instance creation error when creating a\ninstance that belongs to a cluster. Troves\u0027s Instances API has\nchanged a parameter name from \"net-id\" to \"network_id\", but Clusters\nAPI and python-troveclient still uses the \"net-id\" parameter name.\nWe need to fix it.\n\n[1]: https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372\n\nTask: 44967\nStory: 2009958\nChange-Id: I7ee128afad80b92bdeabb45f719d19a99badd8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e518a5513927bd96ea8e8d6c6395282ea626848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e518a5513927bd96ea8e8d6c6395282ea626848"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
