)]}'
{"id":"openstack%2Ftrove~838175","triplet_id":"openstack%2Ftrove~master~I759d482430eaf0e9304986c9fad047cbd44c09ca","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-04-18 01:32:04.000000000","reason":"wu.chunyang replied on the change"}},"hashtags":[],"change_id":"I759d482430eaf0e9304986c9fad047cbd44c09ca","subject":"Add image build test on Centos8 stream","status":"MERGED","created":"2022-04-15 13:07:52.000000000","updated":"2022-04-18 02:06:06.000000000","submitted":"2022-04-18 02:00:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838175","meta_rev_id":"c6daa68181ab25c4d6c0aebe9c38da90a4ca10be","_number":838175,"virtual_id_number":838175,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-18 02:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-18 01:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-18 01:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-15 13:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2233e74c24ca606064dc0b27ba2d527b3bd7aa41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-15 13:07:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f11a972d24e5dd22664cc23fe744b97ff4773ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-15 13:44:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02532e40de3aba4bf5e309810e1f855edb98dc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:45:32.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 22, column 19","accounts_in_message":[],"_revision_number":2},{"id":"0ec7288f18a2929ace24a65359e329129a673e36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-15 13:52:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f77a1de1ffe59bb5fa552bc0a3248a20ec39e911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 14:23:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3151c9854684b0cb08a58d62d1cfc4c : FAILURE in 7m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a78ed01e7584e8aa8d5a42d9547f9c6 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73687f5c165447a7a5e302ae27a3a3a2 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43df10e4b05e417891808152bd21e43e : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d083e41c7e774e56952a341f4d8aaf63 : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c7b337826c455b9410283fbba14c14 : SUCCESS in 7m 13s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60f174e1cedf417d922155a3c5be8fef : SUCCESS in 6m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4e4bf1cccfee46a6b0c31e8427a27912 : FAILURE in 20m 49s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/4655037d1bdc456cb4a5afc454474cfc : SUCCESS in 8m 55s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/7c6feccf64404464be74028704d5c9ac : FAILURE in 2m 28s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/fcc90386c4394e648fda4419982ae376 : FAILURE in 21m 19s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67e2942bf65146ff8e81f9655ec3caf6 : FAILURE in 28m 33s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4dce4b82bb9d48b39c538e1dacd6d695 : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c6fa20df71a4f639158c58dec9e7af40124fb90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 09:36:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7340e7938ec401ad523707a7e764c5fc08681944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 09:50:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773f78fa4bfc4c198a0d5bf5612e7aef : FAILURE in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c8a8b56ff6443aa1857b0052eaae22 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c392128fc7b34995881eeb1121c68231 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e5e787f5a634a04aa25f56da5ec9098 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0597cda6c1a940ef8775049fa0f5ede3 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a008cc3ebc542a0898e7b191b78bec6 : SUCCESS in 5m 57s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/93703b7d5ae1493fa04e99955d7ca8fd : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":4},{"id":"4da696cf1b94eb51ab75e4adab50ab8c2e0b0499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 10:15:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"83ad11b099b641cbbfcfab20dba307be28608c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 10:31:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1446b9a3ba8141d4abd02714a2ff84ad : FAILURE in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374950536c7648c1874a6714a29e3be1 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63ead9cc77934f34ae7665182f346f77 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50c803309aa34cc4ac6d6c502a7a0c14 : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8a54cf9549c48d5a2f854a101f32287 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91e8d09fec540ed9cf7ef702e1a13bc : SUCCESS in 6m 22s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/b8e978683dd245e0a223f17e3b1fe3ed : FAILURE in 2m 38s","accounts_in_message":[],"_revision_number":5},{"id":"a13798fe4db60a0d8c80f7c0c12b3b4676cad7cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 12:07:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8f458d6091045f3e09ddcd2a388b299333983cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 12:22:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e59b473734f458c96754e409c6a5283 : FAILURE in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1aeeb89e0e846eeb03a165aa379e94f : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ec46800192406094f6ae46b9fe275c : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c84002bd3514fe0af46ad2c911cf66f : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38d4ed43222a4eaabae9d7d812d4d13c : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2524f8034fc743fd86c12497064094d9 : SUCCESS in 7m 08s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/cca8c43021114846b7c89a52626c256b : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":6},{"id":"ee4802b19494a9f21575e613e3898a417c745267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 13:02:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c40c0c083fc6aada8e02018ea85cf51b1efbe806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 13:03:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"93693a2056867857eff255c0f0074ce49cddca8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 13:17:45.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e49a26368e724c7a813a1e1232932bf6 : FAILURE in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0f13eb75be4986945e9cd774cf2861 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9837fa5165f24aa2ab4f4a2805325184 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cc379a177074155bd53f6620b4cd344 : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6943872843e40eda14b79ccbcc7dfb2 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df3b38dcfa94befac0850ec39e00650 : SUCCESS in 6m 46s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/41fa8cbfcb14484ba00802e7e6a38bba : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":8},{"id":"74d9e5a78b4e9f55d9a2a138ccacd20de19ee117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 13:23:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1e4ba23354f0cb9dbdbd0feffdeb1cf7f5c76355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 13:36:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2ff40b5f63a445a929ac53cb348c28a : FAILURE in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a62a513a744724b3038c2e6a66511e : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1f0202dfac540ad96595077318e57b4 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9153237e124122beb56707ac9a507d : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bedd66d7750a408791a16db643aea14d : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c83b280004429c84d5a48e597dabbc : SUCCESS in 6m 58s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/c0046efc33014a688e97882d083e4244 : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":9},{"id":"fbc1809fb49771857e70c5adb12a0c8667b05dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 13:44:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"269f1b7a4a94cd631f0e4d977da1116b5f4bab03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 13:59:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc76d80372f843b8a4af3def2eedd534 : FAILURE in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8771b4e1670b47c28944906ab65c78a9 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e333292555e94116931b293893b42715 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b80325e5bdd4479ab11d45411ee37b3 : SUCCESS in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d040f9d479744061a7122eb5ac24e077 : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771b1a6791904fc3bd29535f1330d56f : SUCCESS in 7m 02s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/eeae6bb886474248b5aadbb625f080e5 : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":10},{"id":"cc9f1dc2e4d050033f7ddf6fde0f1a7e0f70d078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 14:17:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6c4396935aa74ad00337461f8d54c75206f69706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 14:37:20.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f886e9c65a74c06805dabcc43e1dd55 : FAILURE in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d161cfb442964c52bc578fb22e912ace : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4cac707f9984b509fba6153dc8cc983 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba465a3c9df145eabf39dbb9ceb37089 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bf20fe5d1f14825b7d6b2b7090fe877 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dad7b2a9dc445d1aa47e832b38bf73a : SUCCESS in 6m 59s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/706ea33d5a974f5cb89a614d137fee4d : FAILURE in 2m 29s","accounts_in_message":[],"_revision_number":11},{"id":"4be8b652b5571c21e4bf531109be348e7ed4fa3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 14:46:32.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5c55f81d57aa6c7050b40457b8e11c9428f2dd7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 14:57:53.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17d62f4d7872449eb823f74007f1285d : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618bd62e22cd49399972bece15d6316d : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adce08ac79f249b2b6f8e7bbcbc2c533 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50392bb35e944468b7bf0115298a5ec2 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4faa362f7244871ad87e208efbd96f5 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db64489fcdf445009f299e07257752ab : SUCCESS in 5m 10s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/f97412e4192a4fdfbce36dfe2d7e80db : FAILURE in 2m 16s","accounts_in_message":[],"_revision_number":12},{"id":"8373a1740b5d7f5628aadda0181c07e4eeb6531b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-16 15:03:42.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e62ec361e7209f0eb36bf80a2cce3223a1da2529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 15:17:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f95d4d1d128a4096a46e5f98dcfe298e : FAILURE in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22010760a97e45c2866380005baf775f : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2845014e42034d87bf49c0e61c228189 : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1b7bfffb5d042279e0ebe30246acc8e : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66855434d2484aba9133dcfd07859fd6 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef10f5e78042422f861713e6851ee63f : SUCCESS in 5m 52s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/96a8e8f02fd0421f8282d94d738c2d7e : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":13},{"id":"1f6742da7935f9c1526cfa1db7a6315843921e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 03:07:23.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7c9b37e97fb17396d17bdafb4d4d6cb7e3255a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 03:21:04.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f6f50284ee5432cb487d4875c0454d6 : FAILURE in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af76f99ade44385abd44d3bea279f72 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1af6f0315d54ab2b06eb6549e02118e : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f750fa6502dc44b5b5037cb31297e00b : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f54dc57bd83430e8440466e462c80c1 : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f4b311bf4841aa8a40b65c2dcda741 : SUCCESS in 6m 21s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/f699d97662af48419611ca384c526bf5 : FAILURE in 2m 43s","accounts_in_message":[],"_revision_number":14},{"id":"ea26ab63e29bce68b5c5af1c88311f352400abdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 03:27:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"85f6467e731069e8c0ed920fb1a04e46bcaecb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 03:46:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3480aadc9ab460db9fc4d4a67477384 : FAILURE in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025fe2e2af7b4023b8ec0ffdcf6f8210 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ce899c61f294ea89c084431ceba3312 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f7953dbc21341d2a25bf9050fc2d68f : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc6a5707412249bdab197cb4b705aeab : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c6bdebcdb343ad8f258bcbb919256c : SUCCESS in 5m 52s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/66eba7ad38bc450a9ff039858dc6b81e : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":15},{"id":"a243f8bdf7ead31ff318389dc3de480a4762edda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 04:04:01.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a896515da8a3ed08c9b69e67ee7499e666362026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 04:04:31.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7783b32591077543f2e3a1118906910dffc0edcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 04:19:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e70e4173a684aba8ba2e4e103a74e43 : FAILURE in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be8ec589b2442ee911af7ad43198a8f : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b65845fe078e4aee904b825670f999d0 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b64f3eb7bee4799bff07f07ccf358a9 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67d666bc441c458493196d15293fe848 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b96c8b07e4e4fda8dddbdcacaf82eb3 : SUCCESS in 6m 00s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/a95adc01c9c14ecc8bd7b30cd730acce : FAILURE in 2m 29s","accounts_in_message":[],"_revision_number":17},{"id":"2e5b35587bf46c8caf0a1e7fca88b59db951f5b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 05:11:22.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"0f9a44104f7dc8131460536dbc169cae1bfb1dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 05:24:11.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/759e6c8e61594b4eaf5783532e574d53 : FAILURE in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3875c9133a403f98cabcda5a5dd229 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea1ca7ffde64aabb1b358ece1d3a93c : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec52837aa309476eafd7a247af709949 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/908dae26e02e44d3a1b0a6f1bb669d2d : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9066fc212e1348b79bd4a8337c4aa1d4 : SUCCESS in 5m 18s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/3d5f3a87c50247408598fe1398bef242 : FAILURE in 2m 22s","accounts_in_message":[],"_revision_number":18},{"id":"e49ed4a2ad8559eceb34adc026c1f637c69cf476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 05:29:37.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f20443194e0a5fb3a937bba5de21276246da46ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 05:43:57.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1f5c7be26b44a94ad93ac17491c24b3 : FAILURE in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b7e44538c34c35b835dab865b66e46 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b25bf417d72444f5ba3019fffc0c9524 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba22ca7ca91b40acbbd2a7a25b141e73 : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea117fd87d3d4983b9b6a78ae45bf7e4 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc26a6067ada4eaaada9a9f07712a3a0 : SUCCESS in 6m 19s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/ff0ad06d23634804ae7d288555aedac9 : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":19},{"id":"98dfae808c420d4628be1e605714b2434fbed01d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 05:56:51.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"c1146fd0ca80af9fb36eda6743729acf92f8cff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 06:07:55.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b68908c0f42d4fe4877c1795a44decdd : FAILURE in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259108e0087843c4866e45a2f52c37f0 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/178e32d9fdf64385a73dd22f4cd70203 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1c6e66f8e4342408cab69037d4e1b80 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08e59c98455e44c194a59c502dd72e86 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffabdb177fd9495786a385a5d42c4292 : SUCCESS in 5m 17s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/3e1e2ddbf2e248ecab2759042e273638 : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":20},{"id":"87b9c4b9496e77ee0c44e2041d7b8ef2f744bab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 06:34:48.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0aaf262b7b4bc862ee7fc573788693a9d76bceee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 06:51:03.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a978aa703d8a49389c4803f50c408488 : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c696f31a5a45b1b183464f58365922 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b10c7e0e175f4edb953c30ef9cf37500 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4121dc90036343c1a2bb424e18205ef7 : SUCCESS in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f8a29a09bd048a8bbab362ae9169c07 : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc1971b2368425d8d1d02fb5a2cded1 : SUCCESS in 6m 54s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/4094301c3cb44dfdad99e6c94d9ad888 : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":21},{"id":"fbc369be5c962eeae0d3377681bd0aa616b6e53a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 08:18:18.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e10903272e87689d3fc1a782d12a8db712932c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 08:31:15.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae671981c7b84f4ca1cadf90771feca5 : FAILURE in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b26434e6770474e8abda7ec8d0a2b53 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86db5665e4494e2ea42c55b6cef5d4a7 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98ef326c0cc74e75a929ce1dade81dc2 : SUCCESS in 7m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/474cf5b1fd6e49e29f83bc2cf95a93c9 : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e25894bc2d24c1287ebfb8144a3b782 : SUCCESS in 6m 05s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/c978693c27fa44f5ae80a177d93f0cff : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":22},{"id":"91c976422cf4926e359a82c5908287a54d4a9023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 08:49:26.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"5fbf22616a7f68fcd41160e5b63911f0a5da1ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-17 08:54:06.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"23df1d22bd5d60f83ef34a4b90dfc4ebad34268e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 09:46:41.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdc481531d394a7c86814b38f8d5b91f : FAILURE in 6m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8388f3cee9472c9350ac8f6ac4c57b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e626e2ac0af34a95a984b245015a709f : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56e0f09761d04e61ad5fbd5698947a83 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e5ca4a8c1d04f49b04c5fc8c88dbb99 : SUCCESS in 15m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc8ce50d95f422f90711619115ccaa4 : SUCCESS in 6m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d729f5f75c62418eb6a3cdfee62c13e9 : SUCCESS in 6m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/aa88b19fc272438598f3d8a278e45893 : FAILURE in 32m 29s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9fac38acdba04b58a197c34f79220aa4 : SUCCESS in 8m 25s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/a276ae0dbdef4ed7b91bd5cd46c9bd7b : SUCCESS in 10m 51s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/73d14ad842e44dba8eb09e69f7606302 : FAILURE in 31m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4144f74c42d1465eb8e824b6769b791b : FAILURE in 47m 07s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/bc11a9b543694e75bdef7ee9cb0cd45c : FAILURE in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d246c6ed00adec68ff7942548ab9f922d3ac135c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-18 01:32:04.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":24},{"id":"358eb43cf09e6404d8cc16402eb9230ead424423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 01:32:51.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"927c5c47d11dd68947d2c1877b77232a0e1bcba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 02:00:45.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aba9a2b19fb4de086d3f18e09e7548e : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/482dc8b8e36f47b0aaed96424f9c6f75 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca5b5b897f4d4f8aa9e706521b9107fe : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444d8387b6374ba7ab550c0cfb4c5349 : SUCCESS in 6m 53s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b8edcf62396e4171a4c8abfd86e4e69d : FAILURE in 26m 42s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"5ed7d5f0404670acfeff98c1990587072ac54b7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 02:00:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":24},{"id":"c6daa68181ab25c4d6c0aebe9c38da90a4ca10be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 02:06:06.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b4bac566cf4c009fb4175e90fce255 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"4d8e55c5fe0cddaa62008c105d37c5349323f366","revisions":{"95464bfd5d6b5fc1329765e99a09e61d204cda55":{"kind":"REWORK","_number":1,"created":"2022-04-15 13:07:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/1"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95464bfd5d6b5fc1329765e99a09e61d204cda55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95464bfd5d6b5fc1329765e99a09e61d204cda55"}]},"branch":"refs/heads/master"},"88caf02aa631defeb613c5edb4ca56e0990b9068":{"kind":"REWORK","_number":2,"created":"2022-04-15 13:44:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/2"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:44:38.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88caf02aa631defeb613c5edb4ca56e0990b9068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88caf02aa631defeb613c5edb4ca56e0990b9068"}]},"branch":"refs/heads/master"},"cee96d3ff92cae0e70e7658b3889d5bfbe366f58":{"kind":"REWORK","_number":3,"created":"2022-04-15 13:52:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/3"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:52:26.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cee96d3ff92cae0e70e7658b3889d5bfbe366f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cee96d3ff92cae0e70e7658b3889d5bfbe366f58"}]},"branch":"refs/heads/master"},"43e1f6e5213992c8abf689651eedf70988ffd5ef":{"kind":"REWORK","_number":4,"created":"2022-04-16 09:36:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/4"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 09:36:10.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43e1f6e5213992c8abf689651eedf70988ffd5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43e1f6e5213992c8abf689651eedf70988ffd5ef"}]},"branch":"refs/heads/master"},"a8c219329d08d444913beff1eb61a9c08e3c4114":{"kind":"REWORK","_number":5,"created":"2022-04-16 10:15:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/5"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 10:15:42.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8c219329d08d444913beff1eb61a9c08e3c4114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8c219329d08d444913beff1eb61a9c08e3c4114"}]},"branch":"refs/heads/master"},"eccc90698a40a0eabc5d72b2d19c302d72ec142f":{"kind":"REWORK","_number":6,"created":"2022-04-16 12:07:58.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/6"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 12:07:42.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eccc90698a40a0eabc5d72b2d19c302d72ec142f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eccc90698a40a0eabc5d72b2d19c302d72ec142f"}]},"branch":"refs/heads/master"},"f0be8a9ce45e42ac02aef0fef92fe94d97e440b5":{"kind":"REWORK","_number":7,"created":"2022-04-16 13:02:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/7"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 13:02:32.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f0be8a9ce45e42ac02aef0fef92fe94d97e440b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f0be8a9ce45e42ac02aef0fef92fe94d97e440b5"}]},"branch":"refs/heads/master"},"b4722064d234b84cefb6ddb3a7d3a212e3aa073c":{"kind":"REWORK","_number":8,"created":"2022-04-16 13:03:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/8"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 13:03:34.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4722064d234b84cefb6ddb3a7d3a212e3aa073c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4722064d234b84cefb6ddb3a7d3a212e3aa073c"}]},"branch":"refs/heads/master"},"720160d95eda5f1ded6330518fce1df60e26f0f3":{"kind":"REWORK","_number":9,"created":"2022-04-16 13:23:00.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/9"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 13:22:41.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/720160d95eda5f1ded6330518fce1df60e26f0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/720160d95eda5f1ded6330518fce1df60e26f0f3"}]},"branch":"refs/heads/master"},"6fbda17d8be50029981d50cdb371b16eb4dabe90":{"kind":"REWORK","_number":10,"created":"2022-04-16 13:44:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/10"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 13:44:22.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fbda17d8be50029981d50cdb371b16eb4dabe90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fbda17d8be50029981d50cdb371b16eb4dabe90"}]},"branch":"refs/heads/master"},"1ed28c19eb63b479a8d0fa3ef211a7ae7d63b8dc":{"kind":"REWORK","_number":11,"created":"2022-04-16 14:17:39.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/11"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 14:17:24.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ed28c19eb63b479a8d0fa3ef211a7ae7d63b8dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ed28c19eb63b479a8d0fa3ef211a7ae7d63b8dc"}]},"branch":"refs/heads/master"},"81ecad6a31f3e4dde48045fc50838a7420020116":{"kind":"REWORK","_number":12,"created":"2022-04-16 14:46:32.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/12"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 14:46:13.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81ecad6a31f3e4dde48045fc50838a7420020116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81ecad6a31f3e4dde48045fc50838a7420020116"}]},"branch":"refs/heads/master"},"35732b149c4d7226f8e49d10ac3e67c23f42bbfc":{"kind":"REWORK","_number":13,"created":"2022-04-16 15:03:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/13"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-16 15:03:28.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35732b149c4d7226f8e49d10ac3e67c23f42bbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35732b149c4d7226f8e49d10ac3e67c23f42bbfc"}]},"branch":"refs/heads/master"},"bd1265fd1efebecf1196f32f97a9b0935ba60a2b":{"kind":"REWORK","_number":14,"created":"2022-04-17 03:07:23.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/14"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 03:07:08.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd1265fd1efebecf1196f32f97a9b0935ba60a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd1265fd1efebecf1196f32f97a9b0935ba60a2b"}]},"branch":"refs/heads/master"},"bdea77ead6b52758cd7cdcc57c7b5cc8cbadae3c":{"kind":"REWORK","_number":15,"created":"2022-04-17 03:27:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/15"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 03:26:49.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdea77ead6b52758cd7cdcc57c7b5cc8cbadae3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdea77ead6b52758cd7cdcc57c7b5cc8cbadae3c"}]},"branch":"refs/heads/master"},"7f5bc096e84270fd3dd93da9453888eb7e3b0563":{"kind":"REWORK","_number":16,"created":"2022-04-17 04:04:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/16"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 04:03:48.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f5bc096e84270fd3dd93da9453888eb7e3b0563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f5bc096e84270fd3dd93da9453888eb7e3b0563"}]},"branch":"refs/heads/master"},"b08e3a0f03228815480a75ac72f260bd5678c94b":{"kind":"REWORK","_number":17,"created":"2022-04-17 04:04:31.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/17"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 04:04:18.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b08e3a0f03228815480a75ac72f260bd5678c94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b08e3a0f03228815480a75ac72f260bd5678c94b"}]},"branch":"refs/heads/master"},"9d424fefebfc7ad75cde2c8f357634da33b419b0":{"kind":"REWORK","_number":18,"created":"2022-04-17 05:11:22.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/18"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 05:11:10.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d424fefebfc7ad75cde2c8f357634da33b419b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d424fefebfc7ad75cde2c8f357634da33b419b0"}]},"branch":"refs/heads/master"},"24293026eb3a68356330525a36344c04750bb55d":{"kind":"REWORK","_number":19,"created":"2022-04-17 05:29:37.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/19"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 05:29:24.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24293026eb3a68356330525a36344c04750bb55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24293026eb3a68356330525a36344c04750bb55d"}]},"branch":"refs/heads/master"},"def6d042bfda7f838bb2c4b5681ce909ba9293c1":{"kind":"REWORK","_number":20,"created":"2022-04-17 05:56:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/20"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 05:56:33.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/def6d042bfda7f838bb2c4b5681ce909ba9293c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/def6d042bfda7f838bb2c4b5681ce909ba9293c1"}]},"branch":"refs/heads/master"},"5c46886abe4ea103fb1db67e930d224546b04151":{"kind":"REWORK","_number":21,"created":"2022-04-17 06:34:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/21"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 06:34:36.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c46886abe4ea103fb1db67e930d224546b04151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c46886abe4ea103fb1db67e930d224546b04151"}]},"branch":"refs/heads/master"},"fd9013dae00f835d0b2ea9da309c055c7e443d48":{"kind":"REWORK","_number":22,"created":"2022-04-17 08:18:18.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/22"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 08:17:59.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd9013dae00f835d0b2ea9da309c055c7e443d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd9013dae00f835d0b2ea9da309c055c7e443d48"}]},"branch":"refs/heads/master"},"358b7646c58c644ef9b0d14285b2211615f43681":{"kind":"REWORK","_number":23,"created":"2022-04-17 08:49:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/23"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 08:49:12.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/358b7646c58c644ef9b0d14285b2211615f43681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/358b7646c58c644ef9b0d14285b2211615f43681"}]},"branch":"refs/heads/master"},"4d8e55c5fe0cddaa62008c105d37c5349323f366":{"kind":"REWORK","_number":24,"created":"2022-04-17 08:54:06.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/838175/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/838175/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/838175/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/838175/24"}}},"commit":{"parents":[{"commit":"a5ddadddeda20211dca1bc7420d65c1adb95be16","subject":"Merge \"WIP: add guest image build check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5ddadddeda20211dca1bc7420d65c1adb95be16"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-15 13:05:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-04-17 08:53:52.000000000","tz":480},"subject":"Add image build test on Centos8 stream","message":"Add image build test on Centos8 stream\n\nbuild image on Centos8 stream\n\nChange-Id: I759d482430eaf0e9304986c9fad047cbd44c09ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d8e55c5fe0cddaa62008c105d37c5349323f366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d8e55c5fe0cddaa62008c105d37c5349323f366"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
