)]}'
{"id":"openstack%2Ftrove~838770","triplet_id":"openstack%2Ftrove~master~Id3234081fd3114e6536358cbdbf94b8280c8bc41","project":"openstack/trove","branch":"master","topic":"story/2009982-45110","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-06-06 14:16:43.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-06-06 13:49:26.000000000","reason":"wu.chunyang replied on the change"}},"hashtags":[],"change_id":"Id3234081fd3114e6536358cbdbf94b8280c8bc41","subject":"Don\u0027t check task_status when creating a cluster","status":"MERGED","created":"2022-04-20 18:23:08.000000000","updated":"2022-06-07 10:14:02.000000000","submitted":"2022-06-06 14:16:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838770-story/2009982-45110","meta_rev_id":"485acd25c1cf2bb3cb2e1971c7c8658d214bdea8","_number":838770,"virtual_id_number":838770,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-06 14:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-06 13:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-06 13:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 20:31:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 22:57:34.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"974bc95f61b8cef9c7eca6c848025eb6d811b74b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-04-20 18:23:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"424ca1a360bc1b2931dd9912fb1871df506f6832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 20:31:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1057914dcbd34fc08b4af9870afd6be7 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210748610bb74ea39a170a206024f578 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf49040245a41ad86a59665a7ad78db : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a3f1073fc3141a7b215c78663020e0b : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/848b93aee79c4f0ca2f57709ff7a1041 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6857833b2ddf442e8edf8532b16cd7a5 : SUCCESS in 5m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8195231b31154ebca2f55ec2168bd1f5 : SUCCESS in 5m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d86771ca38b6473082c17a4412854fbe : FAILURE in 2h 06m 29s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/4669fe3af7b94070953a5b0737dffd4f : FAILURE in 2h 04m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f0505fc3206455da42eac45b31b8532 : FAILURE in 45m 03s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/923e8058bd5d4cc5aa00334ab8dea415 : FAILURE in 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d4d06b7eb1590841e9cd76645c3978fca541082","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-05-25 22:55:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4267cc682ecce1c973bfb552af860aad9fc6ffa","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-05-25 22:57:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d3f863fd183c7e429bb75543986572f7075487d","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-06-03 02:03:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"131f95f3724fb2d29da7a859d1cf153dcd5c8d24","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-06 13:49:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94541590c3b80c42abbf1fa8a7dfa981f70e5be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 13:50:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bd0b73a63866f339cb3bab98a8496beadb68522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 14:16:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908dac29a5cc4be9a4cacba485c7a1bf : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31713f1bfcf6433582a6142fc96de3e8 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5be40e9d48604718bcd5a546b3b6f4bf : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc4b481937246c7bc91720de4550d2a : SUCCESS in 6m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c3cad0dd02de4923b3ac482545b86e83 : FAILURE in 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ee515352af9882792abf21d6e6c11bc7aba4817","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 14:16:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"25e78aee15ae1216f66ab99c7a9b578b9de6e1d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 14:18:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85a070332b3430385e7757ea289f259 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1},{"id":"485acd25c1cf2bb3cb2e1971c7c8658d214bdea8","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-06-07 10:14:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47835cf92b78bad9b762687ecfa722e3db1a1e5a","revisions":{"47835cf92b78bad9b762687ecfa722e3db1a1e5a":{"kind":"REWORK","_number":1,"created":"2022-04-20 18:23:08.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/70/838770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/838770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/838770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/838770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/838770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/838770/1"}}},"commit":{"parents":[{"commit":"a2cab97cc951adbafe10c26a8ebe4f7c8dfd68df","subject":"Merge \"Validates \"network_id\" param creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2cab97cc951adbafe10c26a8ebe4f7c8dfd68df"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-20 18:11:01.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-04-20 18:22:45.000000000","tz":540},"subject":"Don\u0027t check task_status when creating a cluster","message":"Don\u0027t check task_status when creating a cluster\n\nThis PR allows load_simple_instance_addresse to get the ipaddr of an\ninstance from neutron when creating a instance belongs to a cluster by\nskipping checking the task_status.\n\nload_simple_instance_addresses, which is a helper function to load ip\naddr information to Instance instances, currently checks task_status\nbefore getting the information from neutron. However, this logic is\nunnecessary and causes a logic error when instances belong to a\ncluster because the task_status of creating a cluster is determined\nafter initializing a cluster, which happens in the next phase of\nload_simple_instance_addresses.\n\nThe original reason of this code change is that the task status of a\nmariadb cluster hasn\u0027t change from BUILDING to NONE even after\nsuccessfully creating instances belong to the cluster.\n\nTask: 45110\nStory: 2009982\nChange-Id: Id3234081fd3114e6536358cbdbf94b8280c8bc41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/47835cf92b78bad9b762687ecfa722e3db1a1e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/47835cf92b78bad9b762687ecfa722e3db1a1e5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
