)]}'
{"id":"openstack%2Ftrove~841075","triplet_id":"openstack%2Ftrove~master~I2ffe6c14a56a2f9ac259556ba329b148321a9395","project":"openstack/trove","branch":"master","topic":"py36","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2025-07-06 22:24:59.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I2ffe6c14a56a2f9ac259556ba329b148321a9395","subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","status":"ABANDONED","created":"2022-05-09 07:38:43.000000000","updated":"2025-07-09 01:03:44.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f97405b9cd6cc41cfc5b33188e994421b8589b62","_number":841075,"virtual_id_number":841075,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-09 08:16:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-11-02 14:35:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 08:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 14:35:53.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2025-07-06 22:24:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-07-06 22:24:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"}],"messages":[{"id":"0eff86dfeeed8e595aa74137fe3b84d17b9e3de0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-09 07:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44af3ccce465e144b8500cad143dd5913c72b385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 08:16:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/044dbaeb15f34a188b0222dc637aafec : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/101e5e2957b242c6b36dbe335b775e85 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532e1e40e6df44a3a19ed6deeddcc0a3 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67d5f3a017504b0db38c179e618ed647 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c3f56334dc4efd90937973832cb534 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76cddfe3eb8d40a8b9bebcd8650693a3 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0028f9e1056488c9819db0fd9525e0b : SUCCESS in 6m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2c13be5a478542428cc29730538cd352 : FAILURE in 32m 17s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/fa145f404aec4d3489a4cdcf098393d2 : FAILURE in 32m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af51c75e6eb64090af88c1b8c7914cdc : FAILURE in 36m 18s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0c31217edb5c43cbbb013f7f7d21bfdf : FAILURE in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdbcd086343c6ec902b4e6a2d52b1db257a9f788","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-02 14:35:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8079cb6c242db54da313fba551c72e089eae860d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 22:24:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f97405b9cd6cc41cfc5b33188e994421b8589b62","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-09 01:03:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c205898ef33983a909a929a5f47d244c9d1e6bb5","revisions":{"c205898ef33983a909a929a5f47d244c9d1e6bb5":{"kind":"REWORK","_number":1,"created":"2022-05-09 07:38:43.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/75/841075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/841075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/841075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/841075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/841075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/841075/1"}}},"commit":{"parents":[{"commit":"3d0782707bfd94f0a310b09c6794a7bd9b2bf48b","subject":"Merge \"instance model: fix extraction of ip addresses from instance record\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d0782707bfd94f0a310b09c6794a7bd9b2bf48b"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2022-05-09 07:37:50.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2022-05-09 07:38:28.000000000","tz":480},"subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","message":"The Python 3.6 and Python 3.7 Support has been dropped since zed\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I2ffe6c14a56a2f9ac259556ba329b148321a9395\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c205898ef33983a909a929a5f47d244c9d1e6bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c205898ef33983a909a929a5f47d244c9d1e6bb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
