)]}'
{"id":"openstack%2Ftrove~841824","triplet_id":"openstack%2Ftrove~master~Id6a6aae3de17c12001b1e9946a4c0ce28ff29f73","project":"openstack/trove","branch":"master","topic":"story/2009934","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-07-08 03:37:00.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-07-08 03:13:54.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Id6a6aae3de17c12001b1e9946a4c0ce28ff29f73","subject":"Uses userdata instead of personality files","status":"MERGED","created":"2022-05-15 05:18:01.000000000","updated":"2022-07-08 17:11:17.000000000","submitted":"2022-07-08 03:37:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841824-story/2009934","meta_rev_id":"f30d82c6b612cd82d68b8fc7bb8f8dfd7716adf4","_number":841824,"virtual_id_number":841824,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-08 03:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-08 03:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-08 03:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-15 06:06:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 03:04:56.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"4c54a62291b1bb91b1af0628cfc8d8ecf31d05a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-05-15 05:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12b2f46f2912ca4b1c0b8bc45968d39cd6a7bb20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-15 06:06:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/994720f1a4ab4572b51b6fcd6e05c5ee : FAILURE in 7m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab085b350624c59a4dc916728393d92 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6df8f1465a924c2d8c4f20793195594c : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d63137f17624869b8bf4b459fb74057 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e65f4bd64370413d8f14668bcb11bd6a : RETRY_LIMIT in 2m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439a1930388b48fbbe67653167cc8836 : SUCCESS in 5m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f2fd82617c64f47ad37f94516e60d98 : SUCCESS in 5m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b99135ba7b0a4ab496c86ff10abb688a : FAILURE in 28m 10s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/210723287bf5499bb8a322ed32234ad3 : FAILURE in 33m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdf92c89fc6049f2b413a38e1d1e8cf8 : FAILURE in 43m 39s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0982fe5cae0044d9b0e9a51d1485cd2c : FAILURE in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6796280e513dd31c3cff5257005fc92c19e144d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-06-20 00:48:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99fa2ea75e6af6218f853c4c23a07fe7d20c4e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 01:30:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05741db2679641368b1a1bc69ba89ea1 : FAILURE in 6m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9acce01db454fb7909e609bfe4888d6 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/571e878e3035401aa329405b1533c455 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/434ac72e3a4b4ac1b4fb247088f25a5c : SUCCESS in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abd3e230228941d1ae5f888b3ab7b41d : RETRY_LIMIT in 2m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac40f36d55a47c0a25f93da655132f1 : SUCCESS in 6m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41c84caae735458a81b2670410137e8e : SUCCESS in 5m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b4605c7002014644b5bbd956fa865089 : FAILURE in 26m 18s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b6e2b7ebcb1842f48a77e1f3f22f84f8 : FAILURE in 25m 14s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af95dd08914f43ffbd228cf5d38c2bb4 : FAILURE in 34m 41s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fcfdc7970f634ff49e9b57808b787baf : FAILURE in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a86b04fb40869cd8099b256db634a6fc4a27875","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-07 02:59:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"821c63d372a9ff7ce20e4f79ff25daf73d9694d2","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-07 03:04:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"104c35ed3dab5beca65b04118717afa856cb1747","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-08 03:13:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5564f44d8af800a8965d41c214b1c107b62ba6c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 03:14:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b336102e912677cbf198cae2131f7b9676831b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 03:37:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd036e0508004b4e88783d92cfdbb9be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cbd9a0210b467aa7f0e3a45b3481a8 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7cf9984e9244fc282127588ebcb776a : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bc2962c117c4cfda3ef92839e2e1067 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b2e74d29b64e358f22b76b3f4c738f : SUCCESS in 5m 38s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/544130d1c45d49938119dd79e73305a5 : FAILURE in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13256b23f1f38ae3727daebbab0ccdb16a87e097","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 03:37:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a2a5e70421d0670631c763635c316d49fbdde990","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 03:38:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df0cd753d384fedb43a6b98085ce93a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58705a3f31154aae98156d42061a1667 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"f30d82c6b612cd82d68b8fc7bb8f8dfd7716adf4","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-08 17:11:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b5b93c49bcf9325800ccdb01c5e850090c3a184","revisions":{"2f1b38cde387ed86124bafbadb269bcb9f8bf9c1":{"kind":"REWORK","_number":1,"created":"2022-05-15 05:18:01.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/24/841824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/841824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/841824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/841824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/841824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/841824/1"}}},"commit":{"parents":[{"commit":"fd9a789949f3cf6b4ed41630e71a6c006dfb9f20","subject":"Merge \"Adds new configurations for injected file owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd9a789949f3cf6b4ed41630e71a6c006dfb9f20"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-05-15 05:14:45.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-05-15 05:16:23.000000000","tz":540},"subject":"Uses userdata instead of personality files","message":"Uses userdata instead of personality files\n\nThis PR uses userdata instead of personality files because of Nova\u0027s\ndeprecates personality files[1] since the the NovaApi-2.57 microversion.\n\nOriginal problem is guest instance creation failure due to the lack of\nguestagent configuration files because of the recent code chage[2].\n\n[1]: https://review.opendev.org/c/openstack/nova/+/522027\n[2]: https://review.opendev.org/c/openstack/trove/+/823632\n\nTask: 44821\nStory: 2009934\nChange-Id: Id6a6aae3de17c12001b1e9946a4c0ce28ff29f73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f1b38cde387ed86124bafbadb269bcb9f8bf9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f1b38cde387ed86124bafbadb269bcb9f8bf9c1"}]},"branch":"refs/heads/master"},"0b5b93c49bcf9325800ccdb01c5e850090c3a184":{"kind":"REWORK","_number":2,"created":"2022-06-20 00:48:43.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/24/841824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/841824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/841824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/841824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/841824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/841824/2"}}},"commit":{"parents":[{"commit":"fd9a789949f3cf6b4ed41630e71a6c006dfb9f20","subject":"Merge \"Adds new configurations for injected file owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd9a789949f3cf6b4ed41630e71a6c006dfb9f20"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-05-15 05:14:45.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-06-20 00:46:21.000000000","tz":540},"subject":"Uses userdata instead of personality files","message":"Uses userdata instead of personality files\n\nThis PR uses userdata instead of personality files because of Nova\u0027s\ndeprecates personality files[1] since the the NovaApi-2.57 microversion.\n\nOriginal problem is guest instance creation failure due to the lack of\nguestagent configuration files because of the recent code chage[2].\n\n[1]: https://review.opendev.org/c/openstack/nova/+/522027\n[2]: https://review.opendev.org/c/openstack/trove/+/823632\n\nTask: 44821\nStory: 2009934\nChange-Id: Id6a6aae3de17c12001b1e9946a4c0ce28ff29f73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b5b93c49bcf9325800ccdb01c5e850090c3a184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b5b93c49bcf9325800ccdb01c5e850090c3a184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
