)]}'
{"id":"openstack%2Ftrove~845135","triplet_id":"openstack%2Ftrove~master~I3818bf2a26a84320cdb6d643c983173882d0eff2","project":"openstack/trove","branch":"master","topic":"add-trove-ci","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-07-25 11:35:12.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I3818bf2a26a84320cdb6d643c983173882d0eff2","subject":"CI: Add trove CI","status":"ABANDONED","created":"2022-06-08 13:32:49.000000000","updated":"2024-05-10 14:22:29.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93f3864303fe02204268b835392bc015a39ff22e","_number":845135,"virtual_id_number":845135,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-06 15:40:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 14:38:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c0fe5eeafb6da9bd162fb96850c74c6f2c0b6cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-08 13:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5efaa9c349973ecd931173605185e00286df520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-08 13:36:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5779c64e7f075748858493bcf5eff6f58ad85fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:38:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27388fc3ce224b388803043c19dc22bd : FAILURE in 6m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6401532a0f488ca0c90ef04cdb76c3 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97f2e506cd9f4280a27be53e56106f8d : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd9a655992ca4d448eb8cc63091cffa4 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2b16ba05bdd4cef90931c2b95a5cb4a : RETRY_LIMIT in 2m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7541a806733e400994cf00223177f44c : SUCCESS in 6m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e981734d8115479f876233cb54d14cdd : SUCCESS in 5m 54s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6f1a9fe3107f4a74b86f11194da73582 : FAILURE in 39m 37s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/53b8d389a7b740d2a2a2a3ce03f84b80 : FAILURE in 30m 02s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/160f65a4d6884a90a7b448737805d298 : FAILURE in 29m 27s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7167b58d6f244cf0802d46d34e343d52 : FAILURE in 24m 42s (non-voting)\n- trove-ubuntu-tests https://zuul.opendev.org/t/openstack/build/724c57bb84d5461a97315132b10a7a6b : ERROR Unable to find playbook /var/lib/zuul/builds/724c57bb84d5461a97315132b10a7a6b/untrusted/project_5/opendev.org/openstack/trove/trove-ubuntu-guest-image-build in 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4dedb5e080b443415ba72bd5fbc71dedad9b231f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-08 15:13:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"215486c263f6995bb82a08d9c9ac0ce08b97679e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 16:16:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b22ce05cdc746dfb66d5b01bd01ef94 : FAILURE in 6m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc07511b2f244618b02f1feb642dfe36 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/533c77b1891d432f88dfcc737525f6d9 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98b92afef5654beaa05ae9aa9aa7b64d : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5478490fa1174f41abda3c4b6f326dac : RETRY_LIMIT in 2m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46002389e43748c98670650b98676566 : SUCCESS in 6m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a441d16ff14e4477ada3c54c2f0e705d : SUCCESS in 6m 08s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9022e05054924d0b9f586b8001825adc : FAILURE in 22m 53s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2945dc06046641bf92b21482cc6d4c80 : FAILURE in 25m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d282e66273074731bb8f7a7ef7c4aee4 : FAILURE in 39m 21s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1864308086d946f4ae4496017b8b7470 : FAILURE in 30m 37s (non-voting)\n- trove-ubuntu-tests https://zuul.opendev.org/t/openstack/build/5c6dba65e1c944dbafeca6c63abf576d : FAILURE in 18m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"745e051287c48976c1cf401272f47420931148cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-12 13:32:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b6faf07f806fc0f0fff4f499c21493bac3bfa876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-12 14:18:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51899ef9e78a4191b9bda53c2c2ac380 : FAILURE in 8m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d44cd3ae074f1f89d735e9713c40f3 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dde6e38e27da47c5a764e14275419af5 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/489d44d389b1472ab615c5ce32c3e2aa : SUCCESS in 8m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b695474e1c747048a7fc3b9b819e7c9 : RETRY_LIMIT in 2m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a483b9a7259c4033b2b50331d883d996 : SUCCESS in 5m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f651fe84ba748c2ba46c7c5ad2b52c4 : SUCCESS in 6m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dfa8dc17418346fe90327a4741e55946 : FAILURE in 44m 59s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/3f0b4247498e4f19abfa40309c0248b5 : SUCCESS in 8m 54s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/2fa0bfb8155948f088ec762d101c5c45 : SUCCESS in 10m 22s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e2b478a78da048ccae232d01f08e6245 : FAILURE in 25m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c39d9d89aec74782878c9b8a6de25ad0 : FAILURE in 24m 34s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d767407af71941d6a1f05f30c95f35d5 : FAILURE in 42m 26s (non-voting)\n- trove-ubuntu-tests https://zuul.opendev.org/t/openstack/build/2f3269ee5c27479f94bd71108b27ef04 : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ac7bac082b852ccff716f795512b8e506fc3e172","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-20 12:19:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f2d3ab7425dfe7aae548347a568d8ebfe64bb01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 13:00:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe9c8379ddc84141a81734bf132f5bcb : FAILURE in 7m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0985a21e63884d5f83247593aaa64449 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45b47269b04649c494f2aa316b60f0b2 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eeb326f45404fe6947e2015a17d9336 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e9e3a52261a492f80f15c9abc85a3f4 : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab6a169219e41729725c3654ef3bcb3 : SUCCESS in 5m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b2ca97b6e4544513a8e5aad43e8a2880 : SUCCESS in 5m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b0cba4b4a51b43158890481f928721f9 : FAILURE in 30m 00s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bec78962dbad4e819090fb0cd0b45e17 : SUCCESS in 9m 23s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/8874532996eb4a49931df67326ba3f8a : SUCCESS in 12m 15s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/86542d2b1a4d44ba972924515b09f275 : FAILURE in 23m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d823a196af3e460982bd4d399cab4b31 : FAILURE in 24m 43s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/73289535aadc4810a2fe7e924fd01da5 : FAILURE in 30m 12s (non-voting)\n- trove-ubuntu-tests https://zuul.opendev.org/t/openstack/build/677a467fb82547c1a04975b414d9be62 : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"87b40004a46c9255a3fc5530395dec216a84e5b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-20 14:21:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c124f65fc37f8614515ade35646dca55005257da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-20 14:34:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"819be3713c34940dc18eb22279f3ab0b649edf5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 15:02:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/710b4377034f4984a109f29a4b4c96ea : FAILURE in 7m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64452d7327e42d5bb41dbf2ca354ae9 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aab336dff0d34047ba57b4fc6d7fd53a : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9214d06284a4675aadb75da2b3d42bd : SUCCESS in 7m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8809cc8eea8647cda8b012570d20e9c9 : RETRY_LIMIT in 2m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d946b5dec69246bba6da027adf78d527 : SUCCESS in 6m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c16d0c8473b843058daa7c73f26d57bd : SUCCESS in 6m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5af069ae3c7e442a97c505e2e5308e64 : FAILURE in 26m 11s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/eba43dee185a4caf8200863c2b58348e : SUCCESS in 9m 29s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/e1ddbfd8bc154697860dfaab1ea5a6c5 : SUCCESS in 8m 59s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2c5d918604a043fd881513a5bed69961 : FAILURE in 24m 31s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3435f824f7614f2e83c8d0804a1350e9 : FAILURE in 23m 53s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/aa2f6812b0c0415f9d2a7d2f0cafcf97 : ERROR Unable to find playbook /var/lib/zuul/builds/aa2f6812b0c0415f9d2a7d2f0cafcf97/untrusted/project_5/opendev.org/openstack/trove/playbooks/pre.yml in 38s (non-voting)\n- trove-single-ubuntu-tests https://zuul.opendev.org/t/openstack/build/e5a4bca953524643a3f8aecab0f02328 : ERROR Unable to find playbook /var/lib/zuul/builds/e5a4bca953524643a3f8aecab0f02328/untrusted/project_5/opendev.org/openstack/trove/playbooks/pre.yml in 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5b87f41d7b96278a5954569ecea5c470140bf297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-21 09:12:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"566cdc3571f506055be4e6a1f3b759494fdd5df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 09:59:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50d4cf7d1daf4a6a8ea843d4e585e59f : FAILURE in 8m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf0ee698a344f30b868fd5038e6cc9d : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5d87a47c6748d18cd681b49e158fdd : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0afe601615ab46c886a5a4e1fe32b3c3 : SUCCESS in 7m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43b13222b31c471488e212bc5a06b25a : RETRY_LIMIT in 2m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff21c16ede8f486e8dd4cf641b4cde69 : SUCCESS in 8m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06ec577952704e41a239e0df95401a46 : SUCCESS in 8m 01s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/847eeec40bfd4e1bb12add83e2741e7a : FAILURE in 21m 33s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/402a61db7fce4d54b20597c1875e926a : SUCCESS in 9m 08s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/2293e93b84174247ba801b46eaa722fc : SUCCESS in 10m 52s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d275442c0f134d049a62ac48d5cce671 : FAILURE in 29m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1226154650d4c9b80b240de718feb13 : FAILURE in 38m 21s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f08632a0635a44b3a7b5c5d3ece939fd : RETRY_LIMIT in 7m 46s (non-voting)\n- trove-single-ubuntu-tests https://zuul.opendev.org/t/openstack/build/06826aa144a94d398bc438abbea32c84 : RETRY_LIMIT in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"20b49c9454f5351f9920535a8fd3181629ff9e6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-21 10:04:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c38b5bfd029ab128375aee7124c796866139aa9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 10:31:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bef9c5b26d5430bab5f96570f72cccc : FAILURE in 6m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719c5e352247426dbec1555e5be8ad36 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffaba70f3a8346c89967b8774296d2fc : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1b82243fef498a8ae9adede4a4ac20 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f6ee29ba3c6424abc08f2ca888ea89f : RETRY_LIMIT in 2m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a33779e23145fcbae208269b851c27 : SUCCESS in 7m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d0130f62b9154446935ba3ed64b94dfb : SUCCESS in 6m 10s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9952e78715724ba0ad2d62433205180b : SUCCESS in 9m 35s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/4aa67bb0e4344c50a7224e0fd3e9760f : SUCCESS in 10m 37s\n- trove-single-ubuntu-tests https://zuul.opendev.org/t/openstack/build/c0c12368ad1d480384d04cfa6bf07cfd : FAILURE in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d94ce047e26368a86ac822a0859f8157250f00fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-21 14:08:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5b71f094450e02c2a82b1499f913d6e661143527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:43:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3675dba4432d423ca91a3fd1d65e2c00 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a44f3f33ad64d09afaa3d75348aa470 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfdc7e2b5b0344afa06ede0a686d8e65 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5493545194404c80af87f43fe69968c2 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e86d45a42d44c5887688c9d1823e1d9 : RETRY_LIMIT in 2m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75a1d15f205444da5f1673f74fc1a40 : SUCCESS in 7m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/417154e1218041aeb111d0d52a692d48 : SUCCESS in 6m 22s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/aafbfc121f7a413fae60a26f7ef778c1 : SUCCESS in 9m 42s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/38ed3d0ffed04a428c51d19a83adf2ca : SUCCESS in 10m 36s\n- trove-single-ubuntu-tests https://zuul.opendev.org/t/openstack/build/980abf9594ca4a6294dd01201ad18a1b : SUCCESS in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d2827eb4d6f1b438d9941dd66dc9361e423d34e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-21 15:41:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b43c44e0a41471d67a1ce62c8f3f34f43a608b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 15:42:54.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  The key \"parent\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: trove-devstack-base\n      parent: devstack\n      abstract: true\n      timeout: 10800\n      required-projects:\n        - openstack/tempest\n        - openstack/diskimage-builder\n        - openstack/tripleo-image-elements\n        - openstack/trove\n        - openstack/trove-dashboard\n      vars:\n        devstack_services:\n          tempest: true\n          etcd3: false\n          ceilometer-acentral: false\n          ceilometer-acompute: false\n          ceilometer-alarm-evaluator: false\n          ceilometer-alarm-notifier: false\n          ceilometer-anotification: false\n          ceilometer-api: false\n          ceilometer-collector: false\n          q-svc: true\n          q-agt: true\n          q-dhcp: true\n          q-l3: true\n          q-meta: true\n          q-ovn-metadata-agent: false\n          ovn-controller: false\n          ovn-northd: false\n        devstack_plugins:\n          trove: https://opendev.org/openstack/trove\n        # NOTE(zhaochao): we have to override the default settings from the\n        # the base devstack job about \"SERVICE_HOST\" and \"HOST_IP\", as trove\n        # guestagent should use public addresses to access the underlying\n        # components(MQ and Swift), and the previous devstack-gate jobs didn\u0027t\n        # set these variables. However we cannot unset variables in job\n        # definition, so we just set SERVICE_HOST and HOST_IP to empty strings,\n        # and VNCSERVER_PROXYCLIENT_ADDRESS to 127.0.0.1 instead.\n        devstack_localrc:\n          SERVICE_HOST: \u0027\u0027\n          HOST_IP: \u0027\u0027\n          VNCSERVER_PROXYCLIENT_ADDRESS: 127.0.0.1\n          USE_PYTHON3: True\n          Q_AGENT: openvswitch\n          Q_PLUGIN: ml2\n          Q_ML2_TENANT_NETWORK_TYPE: vxlan\n          Q_ML2_PLUGIN_MECHANISM_DRIVERS: openvswitch\n      #pre-run: playbooks/pre.yml\n      parent: grenade\n      #run: playbooks/trove-devstack-base.yaml\n      irrelevant-files:\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^releasenotes/.*$\n        - ^tools/.*$\n        - ^trove/hacking/.*$\n        - ^trove/tests/unittests/.*$\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 2, column 5","accounts_in_message":[],"_revision_number":10},{"id":"ae94f1eaeb07057703879ebf3ddce751f69f0c17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-22 01:01:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"77320f22104daad3e684718093b3a57f619ae14c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 01:33:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff2908dae0d41ab9449bca960cd27f6 : FAILURE in 6m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc0273dd4bd40c5b22a929b8b5fc7dd : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca1515eb980545d0ae49922c72b9cfea : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab94914e5eec466d8af367a10783b237 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/381d60ef8ea04bc8823e279addcb2bf0 : RETRY_LIMIT in 2m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/983f33aef7d549cd997945225d4f585b : SUCCESS in 6m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff2cbed5b5aa45f8a291495de901bff8 : SUCCESS in 5m 39s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bf4b03288dee4951970cc468c83683ca : SUCCESS in 8m 44s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/2142bc21c6e846b49d74b83abcfa7bbc : SUCCESS in 11m 25s\n- trove-single-ubuntu-tests https://zuul.opendev.org/t/openstack/build/a147052129e24095ac1cb7a9cfb0cfb2 : RETRY_LIMIT in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4a0b3863349f3633e0ac376544622b0ca5fcfc49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-22 05:47:25.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9159f0770cc05527d8cd546358d155ecccb33b79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 05:47:48.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  Job trove-single-ubuntu-tests not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        # - openstack-lower-constraints-jobs\n        - openstack-python3-zed-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: false\n          - trove-tox-bandit-baseline:\n              voting: false\n          #- trove-tempest:\n          #    voting: false\n          - trove-ubuntu-guest-image-build:\n              voting: true\n          - trove-centos8s-guest-image-build:\n              voting: true\n          #- trove-tempest-postgres:\n          #    voting: false\n          #- trove-tempest-ipv6-only:\n          #    voting: false\n          #- trove-functional-mysql:\n          #    voting: false\n          - trove-single-ubuntu-tests:\n              voting: false\n      gate:\n        queue: trove\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-bionic\n          - publish-trove-guest-image-ubuntu-bionic-dev\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":12},{"id":"4d9c17c82d82daa284af6f9f629fafddb623a53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-22 06:55:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"43ee8a2ab3bf3ee9c1870030ea5fa0b308e4273d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 07:34:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ed7fab16d5c484da6ea533e88c3dee4 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f93584362c48f6993c1540b7ec35a4 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bab029d97c648a7ae4242c93e9b5b57 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c0546f5f5584d60b24db6304e69b114 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6493e343246845b58e8d95903f01635d : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03765b90411247648bda635cbba3d09f : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ba979073a0d447fa7e5f8c6046129cd : SUCCESS in 6m 05s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/e3ca1612e7c84fe3b7d9e47f870a84dc : SUCCESS in 8m 37s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/c963192d79d2474ebaba5055c5df9988 : SUCCESS in 9m 20s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/97eca912550c4c33894388cfa589e075 : RETRY_LIMIT in 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e480788c5f18a8bbbd4b9689b5f2d87c64ef4b77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-06-22 09:15:25.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"dfad02deadd730133fb0ac51d9957b8e01023066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 09:37:04.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e3a90f283934161b93abaf0ae580bc9 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a22c96b9cf49ef903d744305480118 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3112076cef664bdeb590d6c42f8b6c06 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69feda8d2d664cca8ca527ad200a75cf : SUCCESS in 8m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63ba030b12be4f9886a71bece0102a08 : RETRY_LIMIT in 2m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8dfe62397540839f163878eff78d84 : SUCCESS in 7m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b81c629b1e164cdcb1957d7aeefacd6a : SUCCESS in 7m 22s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/da38d71aab35498682cb432f24171121 : SUCCESS in 9m 36s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/c9daf643647b4f68baa5160357018831 : SUCCESS in 12m 24s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/81480e70859044fd9fcbc8b81d97ad74 : SUCCESS in 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a659b2c7ab5e18736b297e8815a3e0534322a1c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-13 02:43:42.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"621eb03d8c50789a8c7d9f172dff2bb06850ccd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 03:17:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ad752273d1545b8a1dcb057ea4953b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f01577846b644a14abf4ae78fe724221 : FAILURE in 7m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcbb1d711a1542cca6ed338d5f6435be : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/813d732b016f4dcf9d9b708eddb8fa8e : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c81516f8da2434d8a84a1ba30dbcb28 : SUCCESS in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/607730540cdc49e0ab0d363a91c84ffe : RETRY_LIMIT in 2m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb1e4128f5c42f1b24456a864401546 : SUCCESS in 6m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1cf9ddfad3a47deba2dabc8edfedee0 : SUCCESS in 8m 35s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/6d93132c4a1245b29a77275252496e11 : SUCCESS in 12m 28s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/d34c031250b44ab5a82e6e0de2386e34 : SUCCESS in 14m 17s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/a3cdaa37ac704ef88725efa21952c334 : FAILURE in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f121964599d27d06890eed9f5cd6054dc2214e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-13 03:31:38.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"5176125f2c9b485ebf0d89c5624be20b305e8da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 04:00:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee7872650ee48f3a927bf80ee145c0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c294a2efe0ac4d90bcfdfd9aa69ede5b : FAILURE in 7m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b55ccbacc7f485daef300e2711806cc : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa4367365f594fecbf02269c8854809b : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dbaa670952f490894b32c3396b3155f : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/886aaaa5a26a42b6bb6c4e4765a7639b : RETRY_LIMIT in 2m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71eb4ad3f9774e49822f8556d7780dc0 : SUCCESS in 7m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/24179a5563114f67affb877c71de48cd : SUCCESS in 6m 17s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/1b5e3b39dce64934be1080a9253d3ee9 : SUCCESS in 8m 44s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/1604fad761a34eb395cc9364a1c73cf0 : SUCCESS in 11m 41s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/64efc0b31e2c4351a98ad4c83df2bf5e : FAILURE in 25m 50s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"3619db38925bbce14ed099e9d1a809ce21a51a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-13 06:35:10.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f43642440a3b57c5519e3f7e0e6f7bb34f4ed196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-13 06:36:57.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"5399dd4f02af46130e2644b0e7e85092f860e31b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 07:17:07.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7583a84fe3c8466c9bff78f221a761fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08f1261b892b475fad11ff47bb9a2e7a : FAILURE in 6m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf70a52e3da4038b6df8490d5e85426 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f1f50a177864a1698c83c8d75c8da88 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d9201419a8480eb3542c46453ac81d : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1234a950c8a049479f03386e7158ef07 : RETRY_LIMIT in 2m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a9968d390b4884a31e437a1cd10497 : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9e8af932412749b1930b45ac433cb9d6 : SUCCESS in 5m 18s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/967e473b75dc4703a2d76c9a935dd3e4 : SUCCESS in 9m 05s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/4e93e1a5f48342d6836fb14281a12493 : SUCCESS in 11m 00s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/e9e1d640cd054c33966c5079ade44834 : FAILURE in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6e68bff59abf7799af41fbf2323b3a9e507205bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-20 12:02:52.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"55adbc4267ec0629b44e17278738869e42e39069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:53:06.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1364f80bbabb4e62a2b27d4ef4381250\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe86e644494f48ebac87abb249db5e75 : FAILURE in 7m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99614a5d6744e38be40c7b1e069548d : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d067401940be4487b0db9f0b821da895 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d511984658ae42e187d8465c82fd4445 : SUCCESS in 9m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a94d4d14eb84cefa56e9189b5480500 : RETRY_LIMIT in 2m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e195b660b2804020a0ebef07202e13e8 : SUCCESS in 6m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d7ae33885fd4ae5b120592a05184744 : SUCCESS in 8m 42s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/e45da0e8c9984881a8cf28153aac2f50 : SUCCESS in 9m 04s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/c960048cdc5d4666b940177c487e4302 : SUCCESS in 11m 49s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/676c43a3da184783ab082f540395e3c2 : FAILURE in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4fa3626e45eec2828ae6d3b08145aa277d3c2c7e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-25 11:35:12.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9807bbfadc5fbbfd61185dbf255ba1c6383251b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:18:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e616428104294677957f0d5830e88af1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1362abfc0564c64b438622650f692b4 : FAILURE in 6m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5ee9d8fe0846d197aeb6f860fe5562 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c66e8c330b8f48bf83b3699c05fd504d : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/636b4b442f414bb2885cd57a9ebd0f0c : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29a113e400734d91802e390af9e896a7 : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9d63c7a22a4b4aa2282724a7396fdb : SUCCESS in 6m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0495687d5fde4cedbb2398b7b5cafb47 : SUCCESS in 6m 41s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/cce588f8fd55402fa83015e9d28a5aaf : SUCCESS in 9m 33s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/776edd9b314b4ca5a19bdcef9903868d : SUCCESS in 11m 57s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/5fe0102c41874d7fb3b396689c60ee57 : FAILURE in 35m 47s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"2015aa908cab3c50f5caf3f4d0b1c43d6945de8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-26 02:34:20.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"72dfafb715a170e4330d1eb2bb684c43eba3d737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 03:34:50.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1525b7785cc04c08b7aef336331d8140\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7ad64bb033a4308b3efadaa16904972 : FAILURE in 11m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/622c469686254ccb91b4249054bab2d6 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2374613a684f40069c79308d817b8cb5 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/665e8fa0fe6f46e29f8c6b90cfe316e1 : SUCCESS in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bccaa60dec6340a8b50a0f69bea8c94c : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f537bf2e0f49ea89e80ce26c84c3e0 : SUCCESS in 7m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c1cf8b7cbb4c42e4a4d7e6a441a7f051 : SUCCESS in 6m 26s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/e66f8ea097c44aef8724fb239217581e : SUCCESS in 9m 23s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/e1fc63b9f4a34184aa7fe65c0b563148 : SUCCESS in 12m 31s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/5ac9d49f2d124c0495f8b0b722b2aae3 : FAILURE in 49m 16s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d4f8cb7bdee7606be2ec9e7b208f477492fbee96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-26 06:14:35.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"9f8d5dc91be4f6ed57a6b1d20aa69a61b22eecb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 07:13:03.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32367226edb94e35a78de8e9c404e2bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c207e6b6170741e2afdbb3ad08b7b548 : FAILURE in 9m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951b5d58652a4612b4ac59cab5766258 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16b36054ff3d43be87d0ef5e83ea0a2f : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5f76e7d2fb24e5292b86611e37d4e23 : SUCCESS in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bc6e4bb520b469d941f1427c9a33d60 : RETRY_LIMIT in 3m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc84599d9e814de5916e74e0cad29550 : SUCCESS in 8m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7ddcc6f4d14464585b7706e8256f664 : SUCCESS in 8m 19s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/40c5f71ab4e947998aa73612c0b5b4ff : SUCCESS in 8m 55s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/aa47c0c174e0415396e5f04e09d565e1 : SUCCESS in 15m 45s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/f0cfe5cb5fa94979a72c85e570d9e22d : FAILURE in 45m 44s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"5e37caac1e6f38b6c4499710b79216110c979b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-14 13:00:40.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"3017a9e22fe89a2c7fbaed911e1c4717c826aea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 14:22:35.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81421ea73db54e0891f237fea4357896\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2209797d8e644028e2034a0cff5bb24 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca7d2b2621849009e14981dfc57b928 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13754d47e49d4e8099ecd4b488540d4f : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9adc522ee164a459f728f4c37995b8c : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b1e7eb7b0834a87b13dd9da70014e5f : RETRY_LIMIT in 2m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5f627fd7684781af008c398e998ae7 : SUCCESS in 5m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2731a470d0254648adf9e7dce692d787 : SUCCESS in 5m 45s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/ddcf18406a8c428caf0f70ef2bdaa094 : SUCCESS in 8m 46s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/902c9c19c0124decb662f8162ddcdacd : SUCCESS in 1h 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7f33b3872cb1796741391b8b82ff5b7b328da005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-15 02:48:09.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"13b3b0658c2eb690f114c91c633f3c86fd1ff0d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 04:07:28.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874c6b4951e44b20b25c6b198b840bf0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72e6962d0b04784916e0c5f3aca04c0 : FAILURE in 10m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/228e6b9d2d8f43be9e64bd694edf1847 : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f8162b1a3f6453b921c676ad1ed70e7 : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5233f6b374e5419fa4a581721d3891de : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b7574f508184d59b8416deba40d129b : RETRY_LIMIT in 2m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0186d87a04a464380346bb68c9565bd : SUCCESS in 8m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c841c787830a41789e3b0fe61e4dee4e : SUCCESS in 6m 19s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/95fb466beaa04056ae9d55c28e9818c8 : SUCCESS in 8m 45s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/ffe1109a4e754571acb423b8faece8e7 : POST_FAILURE in 1h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"1755df93814eb65364188327a4e271dcc5cd1c8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-01 14:50:04.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"f2e3cfb0faa561cec1e5be1c9df9281c06d438c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-01 14:52:38.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"091c9d62c8ad5961e1d08bfd90224ed90b7fa25d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-01 14:59:03.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"036cdbc3f35ffe3e1ab94e39f3a7eaaf73ed94c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 15:38:09.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23fabef8f53c42fcb26bc0e3a8c52168\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6775c32c5d48c78415eefb108a59ae : FAILURE in 8m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b1116a235f4c26a2226fc64170086b : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b62ca33a056458eb07d82f6f877285d : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c2af8c1520148549347007f9d70c748 : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1871af4d5dd34c179fcb695bdc7a6865 : RETRY_LIMIT in 2m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b6f4ddd63d4bed8f0647f56c88ab7e : SUCCESS in 6m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb57076544134adc9cafeb4ad0e23d9b : SUCCESS in 6m 45s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/32ab4a0e012b43738d1318d8722c2860 : SUCCESS in 10m 44s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/0da47052cea44655a206e8924aaa7289 : FAILURE in 36m 50s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"eaf38b3961c8df0b95c2df7a66a566d9d8f73d3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-02 10:35:46.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"7fdbb90234b42ca5f06c09adf06e0cf169acd44c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:14:54.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1848102ffc9c4c1493c4c9ee523e617d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9caa883760a4d739f47f85b19acd389 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5937951785ea479da17ca7b7420e6b28 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fab87abb9e754a6bb78e4ad9364b3248 : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d78d765bd1c44bf99f9b975b83889fe : SUCCESS in 8m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32c24ee825e049b9a437d12d974704fc : RETRY_LIMIT in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d364974c34f5477ebb4fae9709b2492e : SUCCESS in 7m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2ac21cd7074044dbb762f1ced10179f3 : SUCCESS in 6m 56s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/d1c671793f4b42b2ba58e3c91f5a649b : SUCCESS in 9m 15s\n- trove-single-node-ubuntu-grenade https://zuul.opendev.org/t/openstack/build/af9051303e5e4cd6949189171e1a6451 : FAILURE in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"fe615bcccef102f16166b1ea4adabd4ed08acd96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-05 06:16:28.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"ff329c15bafb9e8d0c68f739ccc83d1c24965c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 07:12:38.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f7c738f6bf4ae49c8df1a411009fad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12bdded8ab6a465d9d5a822115c03971 : FAILURE in 6m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c33f90bb46e0443bbe34368399608aed : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7e4feaea5c64e80ab60b2b6a8bc85cc : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db9f379914a6446ab28cf178f0130a9e : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b694138f8d1474ea5ea92c919ee8ebf : RETRY_LIMIT in 2m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4eab26698849109c018845ce7eb128 : SUCCESS in 5m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46c0f9f2dc70495fb43e01b8c88ccee5 : SUCCESS in 8m 02s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/4616d125559f4fc3afcc6c1a932acbad : SUCCESS in 13m 20s\n- trove-single-node-ubuntu-devstack https://zuul.opendev.org/t/openstack/build/a0804b938e1e4d20b0d4d95ced8fc7a8 : FAILURE in 49m 25s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"c6dc0e760101d6dbfc6bd3168763752dd318fcfb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-05 08:25:57.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"939eac32b6e2c62fa731a3a3e7ba72c46a581b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 09:01:13.000000000","message":"Patch Set 28:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4d2422a663426fa8a287903a2c3267\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f0765b68c5a47088966b03ee199107d : FAILURE in 6m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14a1d7e83724fdd93ce455b3a1fa151 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73e92ec2050e47f0a6a6bb2597127b83 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f623b3608dc54c0a99a4db4f66389166 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04468c8b641341f89034babe9668f487 : RETRY_LIMIT in 2m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddebaa63ff084027968bf7a76b118b25 : SUCCESS in 5m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b602bc98df8542518373e6c4b59d642a : SUCCESS in 5m 55s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bf4dd524acfe4a1a9e7ead5a759a80fd : SUCCESS in 8m 58s\n- trove-single-node-ubuntu-devstack https://zuul.opendev.org/t/openstack/build/103e72dca62c4509b5fc5b8ffe9349af : FAILURE in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"70b26c4820ded60a9d135087dced1bb819bba709","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-05 15:28:20.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"ea13e92345ef103f84cd8edfb2be152c293c325f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 16:10:08.000000000","message":"Patch Set 28:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8abbcb6e501b4782a59a0fefb293e5d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17ea2a817ce94f2eb027090f0db33543 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f8a34629dc4d569aa3b598ee7acfd4 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea54cc9202e469daa8459000ac8934d : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a80420ff271a4fc8a783d9548dc1b048 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e0aedf6629c4b939a8c95a49f01bef3 : RETRY_LIMIT in 2m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93a7925cbca471fa44ddd0c939e0cb5 : SUCCESS in 5m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/821bfa35c50948dda6f66d15f32b5df2 : SUCCESS in 5m 57s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/4479a7a5860a4df0a6dc4fa3fe630a37 : SUCCESS in 9m 31s\n- trove-single-node-ubuntu-devstack https://zuul.opendev.org/t/openstack/build/b7f85cac4e284533b70bc28d8281ba4c : FAILURE in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"158867ab8709d0ccd1aedabb32582ed65d71da6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-06 15:02:15.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"76693bc1372cff815895649481c1c42e5e560d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:40:54.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/395d826259aa4744ae1025a00028d8f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55f8e1177e0a4caab3981fcc35f02463 : FAILURE in 7m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16262c24b4d4d73b87786e463a54966 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af6dfc181cb249eabc761abb05530f87 : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbfd5a47162740f6ba2313cc3608e755 : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef040e9a7e664030b43d1d4fbe81b40d : RETRY_LIMIT in 2m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2ea8a66002483b8f71493a3fd93963 : SUCCESS in 7m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/954b376e98724f68ad700de7db876413 : SUCCESS in 7m 03s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/04922cb20af246dd995ef3e27e5ee90c : SUCCESS in 8m 46s\n- trove-single-node-ubuntu-devstack https://zuul.opendev.org/t/openstack/build/89a4279f5a9b425dad777e134909a57a : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"93f3864303fe02204268b835392bc015a39ff22e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-10 14:22:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"ccaff635139e24074a689a519d2dcc7f00df57be","revisions":{"1fdc2bb352ea3673df38879a633be6fc841b025f":{"kind":"REWORK","_number":1,"created":"2022-06-08 13:32:49.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/1"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:32:44.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fdc2bb352ea3673df38879a633be6fc841b025f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fdc2bb352ea3673df38879a633be6fc841b025f"}]},"branch":"refs/heads/master"},"819e39fa174f2296d484bd39cd72a1c1a9f702e3":{"kind":"REWORK","_number":2,"created":"2022-06-08 13:36:34.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/2"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:36:09.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/819e39fa174f2296d484bd39cd72a1c1a9f702e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/819e39fa174f2296d484bd39cd72a1c1a9f702e3"}]},"branch":"refs/heads/master"},"9de472c2172e53c5c5b31f45f74c6de5513ea3ac":{"kind":"REWORK","_number":3,"created":"2022-06-08 15:13:17.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/3"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 15:13:02.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9de472c2172e53c5c5b31f45f74c6de5513ea3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9de472c2172e53c5c5b31f45f74c6de5513ea3ac"}]},"branch":"refs/heads/master"},"de9377612d14e6beca092b936b035686fa2f5d22":{"kind":"REWORK","_number":4,"created":"2022-06-12 13:32:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/4"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-12 13:32:39.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de9377612d14e6beca092b936b035686fa2f5d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de9377612d14e6beca092b936b035686fa2f5d22"}]},"branch":"refs/heads/master"},"52b4bd18f57d82625d6c32cf109db148ec81573d":{"kind":"REWORK","_number":5,"created":"2022-06-20 14:21:45.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/5"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-20 14:21:29.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52b4bd18f57d82625d6c32cf109db148ec81573d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52b4bd18f57d82625d6c32cf109db148ec81573d"}]},"branch":"refs/heads/master"},"99cd73c305589fc5ec14e2110075bc2aec924e58":{"kind":"REWORK","_number":6,"created":"2022-06-20 14:34:08.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/6"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-20 14:33:40.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99cd73c305589fc5ec14e2110075bc2aec924e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99cd73c305589fc5ec14e2110075bc2aec924e58"}]},"branch":"refs/heads/master"},"76459348672bbec4f5bcba8b664a7ecdd45376c5":{"kind":"REWORK","_number":7,"created":"2022-06-21 09:12:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/7"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-21 09:12:41.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76459348672bbec4f5bcba8b664a7ecdd45376c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76459348672bbec4f5bcba8b664a7ecdd45376c5"}]},"branch":"refs/heads/master"},"b8a5646ecc58dd5409e6bd7bef82e34533cce2fc":{"kind":"REWORK","_number":8,"created":"2022-06-21 10:04:08.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/8"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-21 10:03:51.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8a5646ecc58dd5409e6bd7bef82e34533cce2fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8a5646ecc58dd5409e6bd7bef82e34533cce2fc"}]},"branch":"refs/heads/master"},"b9cf67169c9b5461a941e8da16d7c0a07275cf6f":{"kind":"REWORK","_number":9,"created":"2022-06-21 14:08:45.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/9"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-21 14:08:01.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9cf67169c9b5461a941e8da16d7c0a07275cf6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9cf67169c9b5461a941e8da16d7c0a07275cf6f"}]},"branch":"refs/heads/master"},"c66d9b6d601411e508761065bd4461af66d9a9a5":{"kind":"REWORK","_number":10,"created":"2022-06-21 15:41:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/10"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-21 15:41:08.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c66d9b6d601411e508761065bd4461af66d9a9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c66d9b6d601411e508761065bd4461af66d9a9a5"}]},"branch":"refs/heads/master"},"013506c41b0bd9245fe5b24210acac96b22ab5a1":{"kind":"REWORK","_number":11,"created":"2022-06-22 01:01:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/11"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-22 01:00:52.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/013506c41b0bd9245fe5b24210acac96b22ab5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/013506c41b0bd9245fe5b24210acac96b22ab5a1"}]},"branch":"refs/heads/master"},"7d321185d88543e8dfb677777b71b455a6d5ddd6":{"kind":"REWORK","_number":12,"created":"2022-06-22 05:47:25.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/12"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-22 05:46:59.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d321185d88543e8dfb677777b71b455a6d5ddd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d321185d88543e8dfb677777b71b455a6d5ddd6"}]},"branch":"refs/heads/master"},"eb0f361a1436ac0b2a48d2e0012c8deea8d62427":{"kind":"REWORK","_number":13,"created":"2022-06-22 06:55:24.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/13"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-22 06:55:08.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eb0f361a1436ac0b2a48d2e0012c8deea8d62427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eb0f361a1436ac0b2a48d2e0012c8deea8d62427"}]},"branch":"refs/heads/master"},"b2fbdcc1747326c1973be529b596f1a092064e06":{"kind":"REWORK","_number":14,"created":"2022-06-22 09:15:25.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/14"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-22 09:15:07.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b2fbdcc1747326c1973be529b596f1a092064e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b2fbdcc1747326c1973be529b596f1a092064e06"}]},"branch":"refs/heads/master"},"152e20b9bfafd44e6665c6813e44c7b3b7099412":{"kind":"REWORK","_number":15,"created":"2022-07-13 02:43:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/15"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-13 02:43:07.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/152e20b9bfafd44e6665c6813e44c7b3b7099412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/152e20b9bfafd44e6665c6813e44c7b3b7099412"}]},"branch":"refs/heads/master"},"791155aa7ec8248faf1e47ea3823fc910d104158":{"kind":"REWORK","_number":16,"created":"2022-07-13 03:31:38.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/16"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-13 03:31:23.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/791155aa7ec8248faf1e47ea3823fc910d104158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/791155aa7ec8248faf1e47ea3823fc910d104158"}]},"branch":"refs/heads/master"},"a8a21f3abf1e34888f8b12f78dba828b4ee034e8":{"kind":"REWORK","_number":17,"created":"2022-07-13 06:35:10.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/17"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-13 06:34:48.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8a21f3abf1e34888f8b12f78dba828b4ee034e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8a21f3abf1e34888f8b12f78dba828b4ee034e8"}]},"branch":"refs/heads/master"},"dd48c8adf2bd9c4accf33f3d595bbd74005badb5":{"kind":"REWORK","_number":18,"created":"2022-07-13 06:36:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/18"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-13 06:36:43.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd48c8adf2bd9c4accf33f3d595bbd74005badb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd48c8adf2bd9c4accf33f3d595bbd74005badb5"}]},"branch":"refs/heads/master"},"2a9e1d34f494fe432c2c6adb3027e4fa4a59c8c8":{"kind":"REWORK","_number":19,"created":"2022-07-20 12:02:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/19"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-20 12:02:38.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2a9e1d34f494fe432c2c6adb3027e4fa4a59c8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2a9e1d34f494fe432c2c6adb3027e4fa4a59c8c8"}]},"branch":"refs/heads/master"},"9ce983e7d06bf47be83590e03bc78b4a5b58fee3":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-07-26 02:34:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/20"}}},"commit":{"parents":[{"commit":"f4783ad4973481ae7136f19716ebbb8a6eba286c","subject":"Merge \"Fix import from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4783ad4973481ae7136f19716ebbb8a6eba286c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-26 02:34:20.000000000","tz":0},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ce983e7d06bf47be83590e03bc78b4a5b58fee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ce983e7d06bf47be83590e03bc78b4a5b58fee3"}]},"branch":"refs/heads/master","description":"Rebase"},"a267b3bc40b472d2e6ea8f6a9d58772b131273f9":{"kind":"REWORK","_number":21,"created":"2022-07-26 06:14:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/21"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-26 06:14:18.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a267b3bc40b472d2e6ea8f6a9d58772b131273f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a267b3bc40b472d2e6ea8f6a9d58772b131273f9"}]},"branch":"refs/heads/master"},"673a460703b8d858ea765d2192f3e0966e35c192":{"kind":"REWORK","_number":22,"created":"2022-08-14 13:00:40.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/22"}}},"commit":{"parents":[{"commit":"06773322c5590c286c9afc251d34bfcb5c8a2005","subject":"Merge \"Fix docker start failed in guest-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06773322c5590c286c9afc251d34bfcb5c8a2005"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-14 12:59:35.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/673a460703b8d858ea765d2192f3e0966e35c192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/673a460703b8d858ea765d2192f3e0966e35c192"}]},"branch":"refs/heads/master"},"b5ceb05cf6d081eec195327c57395dac1a9626b2":{"kind":"REWORK","_number":23,"created":"2022-08-15 02:48:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/23"}}},"commit":{"parents":[{"commit":"06773322c5590c286c9afc251d34bfcb5c8a2005","subject":"Merge \"Fix docker start failed in guest-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06773322c5590c286c9afc251d34bfcb5c8a2005"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-15 02:47:53.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5ceb05cf6d081eec195327c57395dac1a9626b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5ceb05cf6d081eec195327c57395dac1a9626b2"}]},"branch":"refs/heads/master"},"0bd707cd42aa126d50bd929f133ac20e2df549e4":{"kind":"REWORK","_number":24,"created":"2022-09-01 14:50:04.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/24"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-01 14:49:46.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0bd707cd42aa126d50bd929f133ac20e2df549e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0bd707cd42aa126d50bd929f133ac20e2df549e4"}]},"branch":"refs/heads/master"},"88cf275e72d0cd853152b9444f18b14b82eb678a":{"kind":"REWORK","_number":25,"created":"2022-09-01 14:52:38.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/25"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-01 14:52:20.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88cf275e72d0cd853152b9444f18b14b82eb678a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88cf275e72d0cd853152b9444f18b14b82eb678a"}]},"branch":"refs/heads/master"},"54a2dc611d86a747005d620dae8ff95ef12f47dd":{"kind":"REWORK","_number":26,"created":"2022-09-01 14:59:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/26"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-01 14:58:47.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54a2dc611d86a747005d620dae8ff95ef12f47dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54a2dc611d86a747005d620dae8ff95ef12f47dd"}]},"branch":"refs/heads/master"},"26b7ae6b9e46879281337f7e8c28a01b9569d03d":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2022-09-02 10:35:46.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/27"}}},"commit":{"parents":[{"commit":"6e4dd51c02f042efcfd6fb89575f100d44ac0a60","subject":"Rename api_wsgi.py to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e4dd51c02f042efcfd6fb89575f100d44ac0a60"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2022-09-02 10:35:46.000000000","tz":0},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26b7ae6b9e46879281337f7e8c28a01b9569d03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26b7ae6b9e46879281337f7e8c28a01b9569d03d"}]},"branch":"refs/heads/master","description":"Rebase"},"4c7a81694c17a9539e7b0ce3f42c03f4687192f2":{"kind":"REWORK","_number":28,"created":"2022-09-05 06:16:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/28"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-05 06:16:12.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci jobs called trove-ubuntu-tests, The task\nshould deploy trove by devstack and build image later, finally,\nit runs post.yml task.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c7a81694c17a9539e7b0ce3f42c03f4687192f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c7a81694c17a9539e7b0ce3f42c03f4687192f2"}]},"branch":"refs/heads/master"},"ccaff635139e24074a689a519d2dcc7f00df57be":{"kind":"REWORK","_number":29,"created":"2022-09-06 15:02:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/35/845135/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/845135/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/845135/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/845135/29"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-06-08 13:28:36.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-06 14:59:32.000000000","tz":480},"subject":"CI: Add trove CI","message":"CI: Add trove CI\n\nThis PR adds a ci job called trove-single-node-ubuntu-devstack,\nthis test deploys trove by devstack and register images, finally,\nit runs post.yml playbook.\n\nChange-Id: I3818bf2a26a84320cdb6d643c983173882d0eff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ccaff635139e24074a689a519d2dcc7f00df57be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ccaff635139e24074a689a519d2dcc7f00df57be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
