)]}'
{"id":"openstack%2Ftrove~848381","triplet_id":"openstack%2Ftrove~master~I762702654bc4430b8432dd96bf869408b2947ff8","project":"openstack/trove","branch":"master","topic":"story/2010125","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-07-20 12:20:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I762702654bc4430b8432dd96bf869408b2947ff8","subject":"Adds a configuration to use a local docker registry","status":"MERGED","created":"2022-07-01 04:55:41.000000000","updated":"2022-07-20 12:21:22.000000000","submitted":"2022-07-20 12:20:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848381-story/2010125","meta_rev_id":"df5d9d9649baf4e4dd01b674e3e61ea732c8faf2","_number":848381,"virtual_id_number":848381,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 12:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-20 11:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-20 11:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 05:36:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 11:50:44.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"adf0a395ea71701c879a7e0cf26711a15ad637da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-01 04:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c10a465f63fc828f1e1c0c4a39b7294d65e7d761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 05:36:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d01fc0e3f5ed428886b8b13677279eb2 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab2afe9593c489c96fbd7409cfaec80 : FAILURE in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b19bd3a8d724b98a52df7e713c484a2 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf04352c575404ab51fe4a53dc1427b : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c765f8b59474a99a85da02e37ae199e : RETRY_LIMIT in 2m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fdf38eadd2649bb99b3dc61ac4db95f : SUCCESS in 6m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80b10ade2a444f91b8080c61b6043e1f : SUCCESS in 5m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/499a4a3cd3314c5b8c0a30a88710d731 : FAILURE in 25m 44s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d5ed7075a5534a4aab3aae3a76a7ad50 : FAILURE in 22m 54s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11ecf0bc815042ab94cff05795848e7d : FAILURE in 35m 53s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f4e18c39a74647e48ec973e08266438e : FAILURE in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f39dc046ae551745c355a2de985286536763630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-02 08:28:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"032732b044099ae8d6c10b4523122acb77ba63f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 09:03:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4d43e8fd1044738ae288986088da294 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1830926c5274b63bd46815153aa1ef1 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88da8d21c4414b879fe2542977f8d06c : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac91bb5af1384e2ba51dbf91a0c28a85 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52f3c0b7dc8d42678ac801db00e169d3 : RETRY_LIMIT in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f77a2d561614f9abcf806047a5cce6e : SUCCESS in 6m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae44bf5195b646728a1ec83b5d2b439c : SUCCESS in 6m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3d9861e206bf4d3c8a8cb0161e144bec : FAILURE in 21m 46s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0501846bb26d48fdb8aca360e8f406bb : FAILURE in 21m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f5d875697574ae28b4db39337733697 : FAILURE in 27m 55s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/eac532ea702f4eb5900ce81cd7154c37 : FAILURE in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7430f522b3d6ba156774b7dbb518f8c22433afa0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-20 11:50:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69a9adf3e2e60ce897429b40d567e746e7571e7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 11:51:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0658059d17c3f17d02c6c5ebae64fcb48d108a7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:20:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a42bf653d5342fc9bbdcd684fc67f40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55792cbdd19047689469bd0a686045f2 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf238ed20726451c995ee9535304bcee : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51e310e58d3a419a816a0a3e70f9b913 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdae542ab9af4ce290c28b26e769f1a8 : SUCCESS in 7m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/db3751bef7a148e296c86158a462685e : FAILURE in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65ed4b3f5c04c9c56b22263ca4e7018cd164a992","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:20:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"df5d9d9649baf4e4dd01b674e3e61ea732c8faf2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:21:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4a046156aa4c5395a8424572bd89a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6d51335b4b4394930b7116e4c83651 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"893c4b7aad773d08ac578bfd403df7fe28bbe4a5","revisions":{"a76b996af6f73d65b9bd4b4d30327dfe9e3e2fca":{"kind":"REWORK","_number":1,"created":"2022-07-01 04:55:41.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/848381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/848381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/848381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/848381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/848381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/848381/1"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-01 04:42:51.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-01 04:54:49.000000000","tz":540},"subject":"Adds an option to use a local docker registry","message":"Adds an option to use a local docker registry\n\nThis PR adds an option to use a local docker registry for development and\ntest purpose. Using this option in development,\n\n* Developers can test images in local docker registry\n* Developers don\u0027t mind DockerHub\u0027s pull ratelimit\n* Developers can save bandwidth\n* Developers can test database quickly\n\nStory: 2010125\nTask: 44793\nChange-Id: I762702654bc4430b8432dd96bf869408b2947ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a76b996af6f73d65b9bd4b4d30327dfe9e3e2fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a76b996af6f73d65b9bd4b4d30327dfe9e3e2fca"}]},"branch":"refs/heads/master"},"893c4b7aad773d08ac578bfd403df7fe28bbe4a5":{"kind":"REWORK","_number":2,"created":"2022-07-02 08:28:21.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/848381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/848381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/848381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/848381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/848381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/848381/2"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-01 04:42:51.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-02 08:26:55.000000000","tz":540},"subject":"Adds a configuration to use a local docker registry","message":"Adds a configuration to use a local docker registry\n\nThis PR adds an configuration to use a local docker registry for\ndevelopment and test purpose. Using this option in development,\n\n* Developers can test images in local docker registry\n* Developers don\u0027t mind DockerHub\u0027s pull ratelimit\n* Developers can save bandwidth\n* Developers can test database quickly\n\nStory: 2010125\nTask: 44793\nChange-Id: I762702654bc4430b8432dd96bf869408b2947ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/893c4b7aad773d08ac578bfd403df7fe28bbe4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/893c4b7aad773d08ac578bfd403df7fe28bbe4a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
