)]}'
{"id":"openstack%2Ftrove~848382","triplet_id":"openstack%2Ftrove~master~I766b4d61d9eb905cdb4ff46eba0b8cd089ff8b1c","project":"openstack/trove","branch":"master","topic":"story/2010126","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2024-08-16 12:27:24.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-08-16 12:27:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I766b4d61d9eb905cdb4ff46eba0b8cd089ff8b1c","subject":"Adds an option to set kernel parameters in containers","status":"ABANDONED","created":"2022-07-01 05:22:56.000000000","updated":"2024-08-16 12:27:24.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66e6a7362fd489ccfabbb407cc998c976b9b785f","_number":848382,"virtual_id_number":848382,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-02 06:03:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 05:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 06:39:58.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-09-05 02:03:20.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"}],"messages":[{"id":"a701242ca3b7a710d78601f0b5c417ede72c012b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-01 05:22:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0e3f5d23b18530262bfccd5de834f72b2e73961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 05:53:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/244d986247344c45a09e8206e4834e94 : FAILURE in 6m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d029149b974fb7b17c5723e703dc00 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30136283eff942fca9ec1331b0a174fa : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/effd822231e64cdd899e576068fa6406 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/939b53ab6400491095a8514d3313836b : RETRY_LIMIT in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51910ad18e4456c9f17144c3002efff : SUCCESS in 6m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f90e1c5e4b54ee7a6ebb52a6f3d9697 : SUCCESS in 5m 55s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b0af0d9bd5ff43fe8b062c9fd1df0f55 : FAILURE in 25m 16s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ac86bdce5c8f485f91fe159d24059f06 : FAILURE in 28m 19s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7860ecbf5f240cb9e398762a1ddf813 : FAILURE in 27m 56s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/122dbba58f0a490fb690f71e40a8e9a4 : FAILURE in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0133744781565ef7c258ae7c4736c9e61652d21b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-15 09:14:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c8dedc676c0905de4dc09a5479b52ea90fb5440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:56:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fbb23a1aa4f4e13b727a4278a33f219\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6116099989244b38c7c32373f853955 : FAILURE in 6m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32fa0d693ce455c93a1b8fecb49ec05 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec091e60256b469599e76a83638a7c23 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0a3bfbec48d420391df3c9fdd951e89 : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9f7f3a793e248fab00ab0ef5fa559a6 : RETRY_LIMIT in 2m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f0505d0cf5d4ce8a28ca8e38d176b16 : SUCCESS in 7m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d60150844054db392e25cd7f0b427ca : SUCCESS in 5m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b63deac4d15a448bbd4b9b2ae7a16aba : FAILURE in 33m 11s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/40ea0fe40b0444d59f4c21fe28940355 : FAILURE in 33m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe05ba5df7ba4c15b1e1993abf5af275 : FAILURE in 35m 31s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b41f4a6e638546fe9a581d23aa1550aa : FAILURE in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c86f410b9e577491afc1069a73196d394b87115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-02 05:17:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed537b1e0e6597eec140f21a839b8dfc81736f8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-02 05:24:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5f26e64198789971ebc89f5bf9d017da23c3f592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 06:03:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47ce16faa6a148629312d8bd80f2a997\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c37f07ff3ab4ef48d424af2642cf81f : FAILURE in 5m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca58bc7b9744befad271f4d40b1a492 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93375656b3f940c2b6194448f6914dff : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e7549a29863419e94ee3f7e65dae242 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5c1751bbf1b46e38d5a953f0a3e5257 : RETRY_LIMIT in 2m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d220d3b8665040ab9aed6a6e559e31cc : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e37e109f512445bbbcfe39d4251acc5b : SUCCESS in 4m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d97701c6cb0b423786424e451a70783d : SUCCESS in 5m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1338aea139e44b79a5890f78a6227dfa : FAILURE in 32m 24s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/774fc6f2fecd4e97bfc09948edcad445 : FAILURE in 32m 45s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5b321deb7bb4662a1d36a0255ff1776 : FAILURE in 31m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/627e910de3d44574a63ac110bc8deb53 : FAILURE in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b24445f3bf8fc2e53fbaace474b37a072b4c1639","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-02 06:39:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ff59c958808456905d685d16dd95b2da4a09714","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2022-09-05 02:03:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"66e6a7362fd489ccfabbb407cc998c976b9b785f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 12:27:24.000000000","message":"Abandoned\n\nI will abadon this due to duplicated with #926466.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"36b898a9da993d7be30fa758305fd14fb6813508","revisions":{"a1d0b28feadba4ff18297502f5cdb164381a731c":{"kind":"REWORK","_number":1,"created":"2022-07-01 05:22:56.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/82/848382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/848382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/848382/1"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-01 05:08:18.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-01 05:17:53.000000000","tz":540},"subject":"Adds an option to set kernel parameters in containers","message":"Adds an option to set kernel parameters in containers\n\nThis PR adds an option to set kernel parameters in containers to improve\ndatabase container performance using sysctl[1] when starting containers.\nSetting kernel parameters is not required for all databases but, it is\nrecommended to configure kernel parameters in PostgreSQL or other databases.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n\nStory: 2010126\nTask: 45713\nChange-Id: I766b4d61d9eb905cdb4ff46eba0b8cd089ff8b1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1d0b28feadba4ff18297502f5cdb164381a731c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1d0b28feadba4ff18297502f5cdb164381a731c"}]},"branch":"refs/heads/master"},"7736ad9c30320643b5a3b9f4de93414a95322eeb":{"kind":"REWORK","_number":2,"created":"2022-07-15 09:14:23.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/82/848382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/848382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/848382/2"}}},"commit":{"parents":[{"commit":"b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a","subject":"Merge \"Don\u0027t check task_status when creating a cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b36d0b23a2ec66f50ddd5b0e8fdcdd888d19994a"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-01 05:08:18.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:11:50.000000000","tz":540},"subject":"Adds an option to set kernel parameters in containers","message":"Adds an option to set kernel parameters in containers\n\nThis PR adds an option to set kernel parameters in containers to improve\ndatabase container performance using sysctl[1] when starting containers.\nSetting kernel parameters is not required for all databases but, it is\nrecommended to configure kernel parameters in PostgreSQL or other databases.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n\nStory: 2010126\nTask: 45713\nChange-Id: I766b4d61d9eb905cdb4ff46eba0b8cd089ff8b1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7736ad9c30320643b5a3b9f4de93414a95322eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7736ad9c30320643b5a3b9f4de93414a95322eeb"}]},"branch":"refs/heads/master"},"91a7bb52c3f900102d2450dcafbbe1748627f674":{"kind":"REWORK","_number":3,"created":"2022-09-02 05:17:46.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/82/848382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/848382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/848382/3"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-02 05:14:42.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-02 05:14:42.000000000","tz":540},"subject":"Adds an option to set kernel parameters in containers","message":"Adds an option to set kernel parameters in containers\n\nThis PR adds an option to set kernel parameters in containers to\nimprove database container performance using sysctl[1] when starting\ncontainers. Setting kernel parameters is not required for all\ndatabases but, it is recommended to configure kernel parameters in\nPostgreSQL or other databases.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n\nStory: 2010126\nTask: 45713\nChange-Id: I766b4d61d9eb905cdb4ff46eba0b8cd089ff8b1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a7bb52c3f900102d2450dcafbbe1748627f674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a7bb52c3f900102d2450dcafbbe1748627f674"}]},"branch":"refs/heads/master"},"36b898a9da993d7be30fa758305fd14fb6813508":{"kind":"REWORK","_number":4,"created":"2022-09-02 05:24:41.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/82/848382/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/848382/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/848382/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/848382/4"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-02 05:14:42.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-02 05:23:55.000000000","tz":540},"subject":"Adds an option to set kernel parameters in containers","message":"Adds an option to set kernel parameters in containers\n\nThis PR adds an option to set kernel parameters in containers to\nimprove database container performance using sysctl[1] when starting\ncontainers. Setting kernel parameters is not required for all\ndatabases but, it is recommended to configure kernel parameters in\nPostgreSQL or other databases.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n\nStory: 2010126\nTask: 45713\nChange-Id: I766b4d61d9eb905cdb4ff46eba0b8cd089ff8b1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36b898a9da993d7be30fa758305fd14fb6813508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36b898a9da993d7be30fa758305fd14fb6813508"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
