)]}'
{"id":"openstack%2Ftrove~849408","triplet_id":"openstack%2Ftrove~master~I8f157005409fae76c19caadf15c9dd291bd971b4","project":"openstack/trove","branch":"master","topic":"ignore-vscode","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-07-28 03:46:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f157005409fae76c19caadf15c9dd291bd971b4","subject":"Ignore vscode files in git","status":"MERGED","created":"2022-07-11 14:06:32.000000000","updated":"2022-07-28 03:47:25.000000000","submitted":"2022-07-28 03:46:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849408-ignore-vscode","meta_rev_id":"e3e820a808e79d4e2b60f07a2c39b174fb873543","_number":849408,"virtual_id_number":849408,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 03:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-28 02:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 02:57:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-11 14:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 14:28:36.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"53c42155327e920229389412df8f2e6bffcbb023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-11 14:06:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4d9a0f034cff521a552786e0cfd1f05b66a16dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 14:50:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ea12151f09840469a146242ed43f6fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da1c14011cf644f09d0e646351894657 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bdedb134abc4810aa0e3cdaa77121c0 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d7ebb6768d4c18bac03a7982f98f12 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ced2c0e616f4ddc88ef2fe0a029c81a : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a053a49a15441f5be7a500ea2e7b3fb : RETRY_LIMIT in 2m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa5f6e1da2c4f6398138bd910377a95 : SUCCESS in 6m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f9eb373dd054f7e813b570f82449635 : SUCCESS in 7m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6866f583933a443494248b9bb41e79e6 : FAILURE in 28m 00s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7f72b6ed5ebc4fb9ac0b67498d0acd8d : FAILURE in 27m 56s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e344efd9b6b24465adc0c9df6ffe5df0 : FAILURE in 35m 40s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8d0f4f3349f94ea9af7988945a7faec0 : FAILURE in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"706181b022749e1d4e0c383cf88a41a4e9f466d6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-28 02:57:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b3785c3436ab18942d91f50d05fef047e575386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 02:58:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdbc3796d39f5d57dc739781c03c394396deb2ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:46:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eade551551054fc79068894b1534c1f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c1d2a8b6f64273b51a080b72c3533f : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c8b43c9f2014ce5aa8e3f54f6c74e86 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96e88ad4c90d4443809296fbe1d20d65 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74c6ea399c54426ac305e1c8520e51b : SUCCESS in 7m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d2b60ec0a8084771b9bf6ff4493df0c9 : FAILURE in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e46e401e942803b9b450da0e19112bbfd773b013","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:46:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3e820a808e79d4e2b60f07a2c39b174fb873543","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:47:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee7c913ee05416296452b2996d9c251\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378f63cd0ae24d98bf66ae644971c20d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa2425bf06c94d57bbfe2eb0dff33416b8614fdd","revisions":{"aa2425bf06c94d57bbfe2eb0dff33416b8614fdd":{"kind":"REWORK","_number":1,"created":"2022-07-11 14:06:32.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/08/849408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/08/849408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/849408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/849408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/849408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/08/849408/1"}}},"commit":{"parents":[{"commit":"6a4e610bec19c5bac163080b7b7e23b47ee162fd","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a4e610bec19c5bac163080b7b7e23b47ee162fd"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-11 14:05:27.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-11 14:05:27.000000000","tz":480},"subject":"Ignore vscode files in git","message":"Ignore vscode files in git\n\nTrivial Fix\n\nChange-Id: I8f157005409fae76c19caadf15c9dd291bd971b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa2425bf06c94d57bbfe2eb0dff33416b8614fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa2425bf06c94d57bbfe2eb0dff33416b8614fdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
