)]}'
{"id":"openstack%2Ftrove~849972","triplet_id":"openstack%2Ftrove~master~Iaa032fb46ed51b6e416e7d4efdfd272924ba146b","project":"openstack/trove","branch":"master","topic":"story/2010147","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-10-25 09:36:27.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-10-25 06:30:50.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Iaa032fb46ed51b6e416e7d4efdfd272924ba146b","subject":"Fixes the way to check db instance status","status":"MERGED","created":"2022-07-15 09:44:07.000000000","updated":"2023-10-25 14:32:07.000000000","submitted":"2023-10-25 09:36:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849972-story/2010147","meta_rev_id":"cd82999d54af1192ba439afe78dfcdfe66a94712","_number":849972,"virtual_id_number":849972,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-25 09:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-25 06:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-25 06:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 10:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 07:05:39.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"ffd8dd7d1e206f9c1ec9927936389e3cffd61077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-15 09:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2a6c08a97dd4b7f576cb0fed96b09fc23a60386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ffd9749eb44a51a27cb849741af188\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84c2eedebaa747d185cb6f1014d5e314 : FAILURE in 7m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ce469214a14860966da54c3c2b1a47 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c979e8e9d8b84fe58d3a1b1ec2f6bf0c : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/333f38da450e40cd8f8e06106921751f : SUCCESS in 8m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afc17e3eeec142f596052970e2ad5974 : RETRY_LIMIT in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88bf7a5b31204ac5bdbdc6a17b9643ee : SUCCESS in 7m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b2c86356bf194dc9b56c6dd5a7eeb298 : SUCCESS in 6m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5ffe5a1b59ab4762840afd71577c152a : FAILURE in 29m 38s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/8f2c7ec8bea54257a851841a8b311170 : FAILURE in 28m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daa3d3662beb45a786ce4d92f7d82d3a : FAILURE in 33m 36s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e370bd255f5c4453918e51aae1ae3575 : FAILURE in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3863e41236db15d70990679540bf5e6600cf624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-02 06:32:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e837c003b7f79b3c0a8b8386946bbce66756af6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 06:59:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa50ceec5454f859117dece1dadfc34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/204b63b514e548b592eff1fa268d886c : FAILURE in 6m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51976d1709814a5f98ca7c121ed85d05 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ce2ff615ba41a0aadb8e7968ee946e : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37dadf89f91848db816fd5a73af80d8d : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46f9f47d65e042dd9c28492406ea84ac : RETRY_LIMIT in 2m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29c4254600c41c4ad7d9797a42b433c : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65ea4505603742968790876ac5116863 : SUCCESS in 4m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e4c55ec4061e49cd9bd6d25b04eb7d36 : SUCCESS in 5m 59s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b7409df8af6441b4810d138de65a3c12 : FAILURE in 23m 58s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1f8f5c2177be43c28685864edf4a39f4 : FAILURE in 24m 15s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f362d4ba634a4990cda3908b4c1480 : FAILURE in 23m 26s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/46741c577e664b17b2a623d0f75096b3 : FAILURE in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"124795651bc5270a5e85e91656b9aae7ed0840d4","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-02 07:05:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b222381b5a733f31484e19d982395ffefe9233c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-07 02:02:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ebefaec37fc2a64ebf0ac367ac2fdf840617ba95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-16 01:54:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c238d03a7587f3d8bbb8930894946c01aa40bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 03:33:32.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd90d14ab31b469f8cfc8789db451b08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a762f084ead43c3888865675f946ade : FAILURE in 7m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4609c507664811a57002de4e405de4 : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d3591ff709f401fa2e071a9fbbb910f : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f57f483cc26940f793b6e297318475f1 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/655d8b3b131e40a5bc10a1dabce00631 : RETRY_LIMIT in 2m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8333fffdb383429a89816d8ed466b4bf : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d3f6d7811a948c9bfff7d470555c24f : SUCCESS in 5m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d37c445d0004a249fd8b15566c1740c : SUCCESS in 5m 25s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e6500dc0cdf94f4bbedc6726ba782770 : SUCCESS in 1h 37m 24s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e3f9a749009844c999abc13e7ccac18d : SUCCESS in 1h 28m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8657c85141c4c67a41e74edd8afd0db : FAILURE in 30m 32s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5cc50d49465549e7815ff10886732d8d : FAILURE in 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9740fcc214721d75d93c0799c9f8ca90749ab119","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-16 06:18:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d7c28db6f593574dff07c2942062edd72bc08d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:40:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28572168e12e4e19850bc456b4bb91d7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/134de30578594e74adaf4cf63fec0a9a : FAILURE in 7m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8b61f458c64701bd288f8ef24b4532 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d2fc2e662e947739c17b6d5a95a7194 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6fd56b4ca5e45a4b51721deddbe91f7 : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31f84b3129674558929153b5e0232c46 : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7fd9723aef4cf2aac039233b3f71e8 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ca8761c8cbd4998b330740e2d1b94e8 : SUCCESS in 5m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ea2a7b2dc8d49dbb7ae21e033dcf60a : SUCCESS in 5m 37s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/45f1e826b6404b5d9a5fd512fe0e0ebd : SUCCESS in 1h 20m 30s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/dcdcc4ce63d140ceb1b7529585f1cc2c : SUCCESS in 1h 16m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4eae9069feff45ff89bbd9ca02cd7ee0 : FAILURE in 42m 28s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3773f21b1d4d4fc29c2bee06b3a5430a : FAILURE in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c33391d2e732b9ce763bf7e8d29ae43b143871c6","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-09-16 10:54:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0432b5f64e3b9243e0a953029cd69277b7a1c85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 02:50:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"487cfec6fb1769e6a7e2d6d92b39ef04f5554f4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 03:12:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3342681097f743d3b6bb12b1578ca675\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c02a3c3fa2b4ce8809721181f2f5c09 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49cf7561200d41f6b775c67fb96424cf : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66d9f608b597498e8098b245bcb87397 : RETRY_LIMIT in 1m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e45883443ba77434d5b0fe257479fb409c8607bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 05:08:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/111afef2c1ec47a794530f2e2ed515fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f71b7b6d4db24284be61b5d73bedb90a : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/746d267ff0664a5887c78cbac1c6111a : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5971e4259874f449cc2cfb70b4ae02e : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08711c2b71e144a48163d9a8ccd92fee : RETRY_LIMIT in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3666b1ede7c440a6b6d4c8c97d1123f1 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d241c8a4334900b7fd901e273f7aab : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f3f00887e544601a0f107d763dd80aa : SUCCESS in 5m 46s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0a5a0e2caf844521be05036b6add51fc : SUCCESS in 2h 14m 52s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/1882bcc632fa46d6810103bf2c6705a4 : SUCCESS in 1h 29m 40s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d052c77e2a25409cbc38048f2630a278 : SUCCESS in 2h 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"98a0b7f4898f9f8f64d72ccef7102041d4e270c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-25 01:12:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0649b73d95ea0af705bfc66a8f027143a547ba0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 01:34:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d76b29c29941289383f016572b2cd7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7516e2204bc54fdf82fad50de779553e : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f45d6a95959c434cbcd0a3b10bd7bf42 : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/72751cfa6e644fc5a52797cd6f177e19 : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1b228d26f6aa0c427d2f15f1d40cf0477f761e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 03:32:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6870f1e2c2de44798ec09b1fc1800ffe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ae7981e02224a8f8d5785605511d966 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09efef11df9745e6958c63f873c3bc6b : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccbb1e95e47344d4a49e3c48628377e0 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2b52131429845bda210954a1c389dfe : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17105cb73bb8406a84c491b21e6d2df3 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d135d7253746a58397102439aa2f4c : SUCCESS in 4m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f9d17d52c2e3440aaff7e3709ae89b83 : SUCCESS in 4m 44s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3d88c13e220b465f9f55fd4e559e235e : SUCCESS in 2h 19m 02s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/290747c4cb3346c9acc227eb328b197f : SUCCESS in 1h 28m 51s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/909b4c57a8924ac791357b5652a0647b : SUCCESS in 2h 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"25e8af7618732bc32c62ecc221b00b36b0adb02f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-25 06:30:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d42250f94e118a6dd197d39d46f677656dbfbf86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 06:31:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6d443db7b6aaaf6a1890b5957aa09de64627c32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 09:36:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4d485762ab465b88586259764af417\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109fe9d6963746b292591485378ae144 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c69e6412d41e46059b833a30e8ce6a7d : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1b13c0cc47a420b95e54902e724e0bb : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16efd3b548b6427e9ceef45767e5378a : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e76d3880071141e9a3e7a2673ce65447 : SUCCESS in 6m 00s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/60aea72df97748cdb16aca4418e4e4bd : TIMED_OUT in 3h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eb891fb3670f6b4596f76655ca3256568288886d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 09:36:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"5f47d00f21b7a99ad4e1e787f368c346e3f231e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 09:37:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6b28f75e9df459ba8fdb894e0c3103c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6093818b4f4a2294ee688fe7eaf8d2 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9472ad1d9124415c8607c3f9b1fe71bd : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":6},{"id":"83a4ce222495f626486baab10b08e920fb7b4e71","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-10-25 12:33:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd82999d54af1192ba439afe78dfcdfe66a94712","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-25 14:32:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c167159c4a95506f5c72671c9e8d6d59869666cc","revisions":{"40cb427ddf6e10c1b22cea27bc4f818d661f96d2":{"kind":"REWORK","_number":1,"created":"2022-07-15 09:44:07.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/72/849972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/849972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/849972/1"}}},"commit":{"parents":[{"commit":"13db5b37c30cb94ef9d389381ce03af5c8461f4b","subject":"Merge \"Fix error when list database instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13db5b37c30cb94ef9d389381ce03af5c8461f4b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:36:06.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:43:33.000000000","tz":540},"subject":"Fixes the way to check db instance status","message":"Fixes the way to check db instance status\n\nThis PR changes the way to check instance status. Since Victoria, trove\nhas change the db instance status name from `RUNNING` to `HEALTHY`[1].\n\nOriginal problem:\nSome clustering databases like Apache Cassandra fail to update cluster\nstatus because they check db instance status using `RUNNING`.\n\n[1]: https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e\n\nStory: 2010147\nTask: 45791\nChange-Id: Iaa032fb46ed51b6e416e7d4efdfd272924ba146b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40cb427ddf6e10c1b22cea27bc4f818d661f96d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40cb427ddf6e10c1b22cea27bc4f818d661f96d2"}]},"branch":"refs/heads/master"},"9ef5fc0e55ff058f1163253b907fc7e58a366d1e":{"kind":"REWORK","_number":2,"created":"2022-09-02 06:32:49.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/72/849972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/849972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/849972/2"}}},"commit":{"parents":[{"commit":"13db5b37c30cb94ef9d389381ce03af5c8461f4b","subject":"Merge \"Fix error when list database instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13db5b37c30cb94ef9d389381ce03af5c8461f4b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:36:06.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-02 05:39:02.000000000","tz":540},"subject":"Fixes the way to check db instance status","message":"Fixes the way to check db instance status\n\nThis PR changes the way to check instance status. Since Victoria, trove\nhas change the db instance status name from `RUNNING` to `HEALTHY`[1].\n\nOriginal problem:\nSome clustering databases like Apache Cassandra fail to update cluster\nstatus because they check db instance status using `RUNNING`.\n\n[1]: https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e\n\nStory: 2010147\nTask: 45791\nChange-Id: Iaa032fb46ed51b6e416e7d4efdfd272924ba146b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ef5fc0e55ff058f1163253b907fc7e58a366d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ef5fc0e55ff058f1163253b907fc7e58a366d1e"}]},"branch":"refs/heads/master"},"e81dbff4c2db2893f03e16c17834e654cef9d82f":{"kind":"REWORK","_number":3,"created":"2022-09-16 01:54:54.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/72/849972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/849972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/849972/3"}}},"commit":{"parents":[{"commit":"13db5b37c30cb94ef9d389381ce03af5c8461f4b","subject":"Merge \"Fix error when list database instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13db5b37c30cb94ef9d389381ce03af5c8461f4b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:36:06.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-16 01:52:51.000000000","tz":540},"subject":"Fixes the way to check db instance status","message":"Fixes the way to check db instance status\n\nThis PR changes the way to check instance status. Since Victoria, trove\nhas change the db instance status name from `RUNNING` to `HEALTHY`[1].\n\nOriginal problem:\nSome clustering databases like Apache Cassandra fail to update cluster\nstatus because they check db instance status using `RUNNING`.\n\n[1]: https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e\n\nStory: 2010147\nTask: 45791\nChange-Id: Iaa032fb46ed51b6e416e7d4efdfd272924ba146b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e81dbff4c2db2893f03e16c17834e654cef9d82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e81dbff4c2db2893f03e16c17834e654cef9d82f"}]},"branch":"refs/heads/master"},"0fb1b959dd637a90ad8ddd451b518595128fdfa9":{"kind":"REWORK","_number":4,"created":"2022-09-16 06:18:51.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/72/849972/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/849972/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/849972/4"}}},"commit":{"parents":[{"commit":"13db5b37c30cb94ef9d389381ce03af5c8461f4b","subject":"Merge \"Fix error when list database instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13db5b37c30cb94ef9d389381ce03af5c8461f4b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:36:06.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-09-16 06:18:10.000000000","tz":540},"subject":"Fixes the way to check db instance status","message":"Fixes the way to check db instance status\n\nThis PR changes the way to check instance status. Since Victoria, trove\nhas change the db instance status name from `RUNNING` to `HEALTHY`[1].\n\nOriginal problem:\nSome clustering databases like Apache Cassandra fail to update cluster\nstatus because they check db instance status using `RUNNING`.\n\n[1]: https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e\n\nStory: 2010147\nTask: 45791\nChange-Id: Iaa032fb46ed51b6e416e7d4efdfd272924ba146b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0fb1b959dd637a90ad8ddd451b518595128fdfa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0fb1b959dd637a90ad8ddd451b518595128fdfa9"}]},"branch":"refs/heads/master"},"14845f5eb773fb62ac8cfdcf3bbd1b2d5c80afc5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-10-24 02:50:45.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/72/849972/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/849972/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/849972/5"}}},"commit":{"parents":[{"commit":"3bbeeb87e074016db45e6df397b0b379210a5549","subject":"Merge \"Separate backup docker image for each database version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bbeeb87e074016db45e6df397b0b379210a5549"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:36:06.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-10-24 02:50:45.000000000","tz":0},"subject":"Fixes the way to check db instance status","message":"Fixes the way to check db instance status\n\nThis PR changes the way to check instance status. Since Victoria, trove\nhas change the db instance status name from `RUNNING` to `HEALTHY`[1].\n\nOriginal problem:\nSome clustering databases like Apache Cassandra fail to update cluster\nstatus because they check db instance status using `RUNNING`.\n\n[1]: https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e\n\nStory: 2010147\nTask: 45791\nChange-Id: Iaa032fb46ed51b6e416e7d4efdfd272924ba146b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/14845f5eb773fb62ac8cfdcf3bbd1b2d5c80afc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/14845f5eb773fb62ac8cfdcf3bbd1b2d5c80afc5"}]},"branch":"refs/heads/master","description":"Rebase"},"c167159c4a95506f5c72671c9e8d6d59869666cc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-10-25 01:12:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/72/849972/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/849972/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/849972/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/849972/6"}}},"commit":{"parents":[{"commit":"87ba3b787600c11cb24a5c995aa1cce532573ef4","subject":"[CI] fix openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87ba3b787600c11cb24a5c995aa1cce532573ef4"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 09:36:06.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-10-25 01:12:05.000000000","tz":0},"subject":"Fixes the way to check db instance status","message":"Fixes the way to check db instance status\n\nThis PR changes the way to check instance status. Since Victoria, trove\nhas change the db instance status name from `RUNNING` to `HEALTHY`[1].\n\nOriginal problem:\nSome clustering databases like Apache Cassandra fail to update cluster\nstatus because they check db instance status using `RUNNING`.\n\n[1]: https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e\n\nStory: 2010147\nTask: 45791\nChange-Id: Iaa032fb46ed51b6e416e7d4efdfd272924ba146b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c167159c4a95506f5c72671c9e8d6d59869666cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c167159c4a95506f5c72671c9e8d6d59869666cc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
