)]}'
{"id":"openstack%2Ftrove~849984","triplet_id":"openstack%2Ftrove~master~I846f433ff35a9638a070382bf89e02db617dcd3d","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-07-02 13:48:35.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-07-02 10:59:45.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I846f433ff35a9638a070382bf89e02db617dcd3d","subject":"Fixes cluster creation error","status":"MERGED","created":"2022-07-15 10:44:46.000000000","updated":"2023-07-02 13:50:35.000000000","submitted":"2023-07-02 13:48:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849984","meta_rev_id":"258ab4d0661b066d184a87378874e87a7ac2b1ce","_number":849984,"virtual_id_number":849984,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-02 13:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":1,"date":"2023-07-01 23:40:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-02 10:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-02 10:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 11:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 13:46:43.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"ea41873d4e443e304a91c8fdae3b6a980eeef1db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-07-15 10:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e5628dd6c3ae3162f4e7c9991513338d5b44a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 11:29:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7e2840e50a445ba08ade97daecad0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e4a1924583e4bc6a38e1869273282f9 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d89ad72f5e440b8361dcbc8bd08f04 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b99d0924c48c4091969532f2ef28d273 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43ebc9c20d2c45f5a3ffa7e004ae209a : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/713b95d72dc54ebbbc84cec7e60a2c3c : RETRY_LIMIT in 2m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8635e5d731af43cbb53643947852b4ec : SUCCESS in 5m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/068006cdb23c46078318dbdd0bdf277f : SUCCESS in 5m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0f9c243e7bda49a5abd4c4aba7a4bca1 : FAILURE in 29m 14s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/650d8862dd5f47f3b253d2e9904dbc1e : FAILURE in 33m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0493075f60004e54ba52b52e5387d368 : FAILURE in 38m 09s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/88c8e11bee4e466e80d4451e067711ea : FAILURE in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87375d83c475609989f2070a79224c3c77a22f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-11-15 06:19:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d78e251fefcf44c94605e6d555d65ef2d8f122b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 06:39:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/325103521ec44743b7d83d58d08558b8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d04bc69d0c94c49bc5b32e54ad0ccaa : SUCCESS in 12m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e7c72f4318a4194b4f7e53115bfed37 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a56b46f71879ed8660c7ca67f897f317495cc126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 07:34:19.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e800961511a24acf85666bcce528940c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcddd3598e2c4995a038378e18611e7f : FAILURE in 5m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640c56800af64d0b9da0134f21c0c636 : FAILURE in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c2d684bc455419abe4302b0904c68a7 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b4494e8a2dc40858601e2c3782947a9 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3568d9dccc4ed3b6516009c3016ab6 : SUCCESS in 6m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2631bfc269384ecba709250bb11d98f4 : SUCCESS in 5m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/39cefb5ca2e74ac5a55e4fa8977e7886 : SUCCESS in 1h 12m 50s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/06e38721ed6c458c8c1eb4ab275c9c9f : SUCCESS in 1h 11m 56s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a42fb124f80045d796d75e3e32778c99 : FAILURE in 58m 27s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/556f07da0ce3442eb00c4e485fb93792 : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"25f37a95208cadb8b10176e35052cdbad28932bd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-06 13:46:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"97d6091b2686d24f9a2f408a344c4bd5e31615f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-23 17:28:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ae0f1ce48c40dc639f865cdf7757e0273cf26da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:58:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93be5f39787e48dbbf0142e0f6222b85\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081ddc3cd9a1491690dfc902582303aa : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7cd838a2bb6f400d930bce21df4a2487 : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4824ed929d52b500fc90379d2bd29213f8207646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:36:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b1fdafbbf14ac49909bdc20e09e9cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9336481700c54b5a8f1c74e870d52658 : FAILURE in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1b90e203094020ad24d1dad0c70a7e : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1155b563ed64db8a1483c4fb2ed5a98 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8631f9118c44a37821ba3c4ae46438b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3445307e63424ae7be1f314c04237a8c : SUCCESS in 5m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2054d19f5da944889f32b1d0d781d7a9 : SUCCESS in 4m 35s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/63d5559a37874d56ab74acc6665934da : SUCCESS in 1h 38m 55s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7191632abc7d4b3888a12ac68d76e5e9 : SUCCESS in 1h 31m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/813425e516824b89a1fb8555bbe47c73 : TIMED_OUT in 2h 02m 38s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3decf860b2f44a77a48cb63b08dcb76d : FAILURE in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f7b49653827bb5f1b07be0523b82bafbaf49dfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 08:26:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d64cdce207e38a40d747eda753cb39d316acddf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 08:38:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6a2119ad8c4441b84ce3bdcb67caf59\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0529f1ae26a745cd831e14675bf2d101 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/939b97cea181424e894de13bb0251210 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e43e1d2abd64751b15878f7cc1bed47 : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b794c75c3bfcf2737d2091e574545769ea2ddcd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 10:30:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c32aa5d8636947bea646330fcb3a3d82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af4060638dfa4614b3145d0f583b29c1 : FAILURE in 11m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69f2aebb16d4424bc09e835d8196b4f : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cf44491efff44f59fb13edead520430 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d732843ae98341bf8b7ee78756d3d4dd : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75d92ce8e99247ecbefe8135224d01a3 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3ad8be462e47ba95d2d26d025c6f34 : SUCCESS in 11m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc5b25439fdb4475be820f348d8d751e : SUCCESS in 10m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d27f80ad4ccf4fa1aa0dc2a056d44c8b : FAILURE in 1h 21m 15s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/9b7a60a312f94f13a268a7bc37b5c221 : SUCCESS in 2h 03m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aaef30f5042e5b43f0d96d718988c98b1165de59","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 11:06:58.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b8d7b1692834a50ec1e5f4134ea1bd4f5f5a454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 06:26:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"04a161d83ebc0b88a13cc95e76c5ed5944a335bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 06:27:24.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f32d28720d7ee32f8bd826fe2229b1b104b06cd1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 06:28:01.000000000","message":"Topic fiex-cluster-creation-error removed","accounts_in_message":[],"_revision_number":6},{"id":"d5d11ddd0c3c4157d7aa8c51c3d0289c2431cf7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 06:32:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"43b821b339ee5f2a3fdb007e5042f82c9df4441f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 06:33:37.000000000","message":"Topic fiex-cluster-creation-error removed","accounts_in_message":[],"_revision_number":7},{"id":"fa198b6279d1c17120a5631ff4f82635931111bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 06:44:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb16aee646454fc5ba661f33cd76d6a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26bb9a8f3b0f4124a5974907443df164 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c038d9eea73f457ca0a6af07a8653481 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/179e72d672a74a2b9e679b576e19591b : SUCCESS in 7m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e9ff0e41f96a222bf34ccd0b48199f41136505f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 08:37:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/facf289721ff4630bf65e26457aa9a29\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1b579d3d0fb04b7a91b4816924835137 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/891e7fc910ba4b65974c710349ff3257 : FAILURE in 6m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b0782a8c1184863af3066b69f18c9e5 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ca775e646b348978dc80974c2e94e3f : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51966285a0c84ac7bad748fa515dd69d : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e24f9c6c28f48ba92d2a9de54e70d28 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018193ec8d044a649a6053bdcc397e6f : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32096a44866c4f1da00946ec2213aca1 : SUCCESS in 5m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a04c707092524869a2085047f2430658 : SUCCESS in 4m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/358d3eea06384992941a322ac3a83c6e : SUCCESS in 2h 03m 42s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/8304cbb5f7334b6b9fe592785dca3362 : FAILURE in 1h 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f3ba0f6fb4ea57cbe7dc2315ccd1a7c8c60fe96","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-07-01 23:39:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6d2a626fc0d674cdcc475c4645d45b0da8ade0d4","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-07-01 23:40:43.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b90303fd81a8b2828522a98aac134f2d2ae2781b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-02 10:59:45.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2fa22ba9d8028f239aaa6078e3581e36636bf537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 11:00:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e0de28aa7abfd4de981a194d4dbb96f473ef6b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 13:48:35.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7e522f1491c499190770b4be296d374\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91e98b7d8079472281b5507ab9c297d5 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4871338ef2cd4eed8de1292bbd892223 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2001f2f4c5d541de92ddc665204c621e : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8402daa98cb4daf9c4e6f777b1753e0 : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a71cdc6c9fd648da8e0b45e07d112d7c : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad117772e22453995d2cb97428c6c08 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a33c4492e1d4717abd32b0194da9e9e : SUCCESS in 5m 46s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0c007c34e6b54576957ed54f855e8580 : SUCCESS in 2h 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f19d467b52537e42580bab9a306b3375e73f3e8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 13:48:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"258ab4d0661b066d184a87378874e87a7ac2b1ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 13:50:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db03ff7ceb694c0cac079e449bda0c51\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3a3c3cbc232f460fbcb04efe5572ff44 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5be7851cd4441d9aee9b69615f13d0 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c2dce85f2664d73b8d76a371a95f74b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0d4a24c71dfddeafd9aca462ba7e70d0dada2365","revisions":{"5de2921e54b63807e009bd2a99f59dfba389f545":{"kind":"REWORK","_number":1,"created":"2022-07-15 10:44:46.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/84/849984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/1"}}},"commit":{"parents":[{"commit":"13db5b37c30cb94ef9d389381ce03af5c8461f4b","subject":"Merge \"Fix error when list database instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13db5b37c30cb94ef9d389381ce03af5c8461f4b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:44:13.000000000","tz":540},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error and adds\nunit tests.\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5de2921e54b63807e009bd2a99f59dfba389f545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5de2921e54b63807e009bd2a99f59dfba389f545"}]},"branch":"refs/heads/master"},"b3c592e4de8c26e2cb03aaf4a0350bf8474f5ca8":{"kind":"REWORK","_number":2,"created":"2022-11-15 06:19:32.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/84/849984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/2"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-11-15 06:18:54.000000000","tz":540},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error and adds\nunit tests.\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3c592e4de8c26e2cb03aaf4a0350bf8474f5ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3c592e4de8c26e2cb03aaf4a0350bf8474f5ca8"}]},"branch":"refs/heads/master"},"0fce61e85e1e0f858b0f8473733e059c82abb521":{"kind":"REWORK","_number":3,"created":"2023-01-23 17:28:25.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/84/849984/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/3"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-23 17:27:29.000000000","tz":540},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error and adds\nunit tests.\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0fce61e85e1e0f858b0f8473733e059c82abb521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0fce61e85e1e0f858b0f8473733e059c82abb521"}]},"branch":"refs/heads/master"},"8d46d1d2e9b53cd39bb2c3cff401d5abc840206c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-29 08:26:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/849984/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/4"}}},"commit":{"parents":[{"commit":"7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 08:26:09.000000000","tz":0},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error and adds\nunit tests.\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d46d1d2e9b53cd39bb2c3cff401d5abc840206c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d46d1d2e9b53cd39bb2c3cff401d5abc840206c"}]},"branch":"refs/heads/master","description":"Rebase"},"7f42a03963b4c0f52fb1df7de0bff8c33d5ad329":{"kind":"REWORK","_number":5,"created":"2023-06-30 06:26:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/849984/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/5"}}},"commit":{"parents":[{"commit":"7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-30 06:25:58.000000000","tz":480},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error\nand enables flake8-f821 rule checks.\nF821 rule checks \"undefined name\"\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\nCo-Authored-By: wuchunyang \u003cwchy1001@gmail.com\u003e\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f42a03963b4c0f52fb1df7de0bff8c33d5ad329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f42a03963b4c0f52fb1df7de0bff8c33d5ad329"}]},"branch":"refs/heads/master"},"9d5691355e4b206ad4a167d7558321fbc0b12f2f":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-06-30 06:27:24.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/849984/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/6"}}},"commit":{"parents":[{"commit":"7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-30 06:26:40.000000000","tz":480},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error\nand enables flake8-f821 rule checks.\nF821 rule checks \"undefined name\"\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d5691355e4b206ad4a167d7558321fbc0b12f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d5691355e4b206ad4a167d7558321fbc0b12f2f"}]},"branch":"refs/heads/master"},"0d4a24c71dfddeafd9aca462ba7e70d0dada2365":{"kind":"REWORK","_number":7,"created":"2023-06-30 06:32:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/849984/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/849984/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/849984/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/849984/7"}}},"commit":{"parents":[{"commit":"7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-07-15 10:26:57.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-30 06:32:32.000000000","tz":480},"subject":"Fixes cluster creation error","message":"Fixes cluster creation error\n\nThis PR adds the import statement to fix cluster creation error\nand enables flake8-f821 rule checks.\nF821 rule checks \"undefined name\"\n\nOriginal problem:\nwe have currently got `NameError: name \u0027common_glance\u0027 is not defined`\nerror when creating a cluster because I failed to edit the patch[1].\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\n[1]: https://opendev.org/openstack/trove/commit/2f52b2fddbd30dfa68d83e3da32030fa07940ea3\n\nStory: 2010151\nTask: 45797\nChange-Id: I846f433ff35a9638a070382bf89e02db617dcd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d4a24c71dfddeafd9aca462ba7e70d0dada2365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d4a24c71dfddeafd9aca462ba7e70d0dada2365"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
