)]}'
{"id":"openstack%2Ftrove~850353","triplet_id":"openstack%2Ftrove~stable%2Fyoga~I7333a75e8855501f8a89547b93e65157f9f8afb6","project":"openstack/trove","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-07-28 03:46:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7333a75e8855501f8a89547b93e65157f9f8afb6","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","status":"MERGED","created":"2022-07-19 14:19:33.000000000","updated":"2022-07-28 03:47:09.000000000","submitted":"2022-07-28 03:46:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850353","meta_rev_id":"e65ddc8923708eda3bb2530666e8088e04f02950","_number":850353,"virtual_id_number":850353,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 03:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-28 02:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 02:56:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 14:26:20.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2022-07-19 15:11:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3b7e3ff07c0c2aea509f4fae5f07b5f38108790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-19 14:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c37a6d1f2f356a63dd2b6498a20b49ab2fc89a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 15:11:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee041585bb8f442aa02f36780a7ff9ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f10ee8f882425ea54859e93083e0cc : FAILURE in 9m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeebc38e7544481abcfd2b91e7ebae7b : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/411c2d632d6b4bfb9cd9250af088a7fb : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbf98cb370254c79ad4a8c090521729f : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0170fbbf0fea4ae892d6fe91ab5d4992 : SUCCESS in 8m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2b239b14fc34d38b155f489992659ff : SUCCESS in 8m 01s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/49dc1dfa915b47019b248dee8ae20367 : FAILURE in 49m 10s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/f16af54665e343e6ad5d605f70320e35 : FAILURE in 47m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77e52e5fc7f04bda8980a596b9a1dc55 : FAILURE in 47m 04s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7a446864c2e04fc98e7ff093599e601d : FAILURE in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6975a1158c07cc3114ad6d20873eb257e1a5a1e0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-28 02:56:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46df1814abc654f1a3f6a2100ed6e810756cdbd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 02:57:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c86a7a309e57a02b853aa7c001be599d1ee9020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:46:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f39cd1057ead4a8fafc6e3e4fae45f9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73de33d3f06442a9906aa28cc2b89451 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e38f02ade0144af599213e5ed2b63f71 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d661a3e9df394ceb9ff6442ca6f744eb : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7c9d3b799d45388276d370a1ea3d5f : SUCCESS in 10m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0c3d7a8726424645b0712117fd880813 : FAILURE in 47m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bfd63417913ea6aee799a6657fb6942187be8d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:46:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e65ddc8923708eda3bb2530666e8088e04f02950","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:47:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f737e405e249098ceeebb5a6b10e0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e6bcc412d548d289f5e57b566d2bcb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e30d4c93159b55f5e54f595cc057d9b3dd4ae72d","revisions":{"e30d4c93159b55f5e54f595cc057d9b3dd4ae72d":{"kind":"REWORK","_number":1,"created":"2022-07-19 14:19:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/53/850353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/850353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/850353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/850353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/850353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/850353/1"}}},"commit":{"parents":[{"commit":"94c7abd5d5b73ee395f78eda987e06d12a2ed7b9","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/yoga\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94c7abd5d5b73ee395f78eda987e06d12a2ed7b9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:15:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:15:31.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","message":"Update TOX_CONSTRAINTS_FILE for stable/yoga\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nStory: #2010137\nTask: #45767\nChange-Id: I7333a75e8855501f8a89547b93e65157f9f8afb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e30d4c93159b55f5e54f595cc057d9b3dd4ae72d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e30d4c93159b55f5e54f595cc057d9b3dd4ae72d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
