)]}'
{"id":"openstack%2Ftrove~850370","triplet_id":"openstack%2Ftrove~stable%2Fxena~I7333a75e8855501f8a89547b93e65157f9f8afb6","project":"openstack/trove","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-07-28 03:46:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7333a75e8855501f8a89547b93e65157f9f8afb6","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","status":"MERGED","created":"2022-07-19 14:21:35.000000000","updated":"2022-07-28 03:47:18.000000000","submitted":"2022-07-28 03:46:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850370","cherry_pick_of_change":850353,"cherry_pick_of_patch_set":1,"meta_rev_id":"3e705d4042c520f00ef06190f93bb5eb37e6f33f","_number":850370,"virtual_id_number":850370,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 03:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-28 02:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 02:57:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 14:26:47.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2022-07-19 15:00:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42d8a5d10dad25aa8c31a1170f30c5234ecae6e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-19 14:21:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"a10ffbf831eeacf6f1b960c164e6ce27e1263e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-19 14:22:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ca81087ac6c907c80bbcefc5f30e892c9934fbc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-19 14:24:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e5d56456e8d117ee77f2e8cab8a7bbe98f396682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 15:00:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313c1f17efab4c6fa34dfb9bb49630b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f08d64b03da49f79b0ec77ce609a7a3 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b263c854f7d433fbcdac09b7b8ed001 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61153aec783e4ed6946b8ab6bb2f6407 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8696916ac954498188a817d4c6f0b19a : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66ba828637de43429fa70946896a4785 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de31be81147a4ebd9301c4cfe3ea4a58 : SUCCESS in 7m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ad1d929b39e49fa87462bee2121f52e : SUCCESS in 5m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1dd7e5c2cd8d40aca8f4e6a1dc53ec5c : FAILURE in 32m 52s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b993383f9f5f47d18d7c2d389cf2f230 : FAILURE in 25m 13s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885047a1a34b43a488793860a42dde88 : FAILURE in 34m 03s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/39d4f12830e4402a8d6d582e8e4b0267 : FAILURE in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d80e244d20037950269973c7ef65c28de59a5b32","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-28 02:57:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b6a34add253ece06eb3a5c4c6d11780ba2248dac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 02:57:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"55c73edf25534e087ae332441f1f42355ab0b0bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:46:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d408c31c9e6440c28a1784e1377a4292\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf5870060f348949efb0f4feaa2cf0e : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f25ffba669e4135a8cab16f87b433d3 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/188e7995a52948d3b131d98693791e03 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f347a2b6984ad2a99cf43a94bfffb8 : SUCCESS in 9m 09s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e9d5a81f81e948999df42a46aa5e6f27 : FAILURE in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9129c962c9d346da239e06ef35dd8f4f0ab5bb70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:46:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3e705d4042c520f00ef06190f93bb5eb37e6f33f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 03:47:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef154cb5a0764e24a95323ea272a26e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b224b91964b7405e968b8de40ad1f8ae : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"425f065d8896a7aff0ab82316bf4865cece9205c","revisions":{"37df560c2ba6c6ce0f775aec519e065833621125":{"kind":"REWORK","_number":1,"created":"2022-07-19 14:21:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/850370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/850370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/850370/1"}}},"commit":{"parents":[{"commit":"f3864fc262121881bd676f832c363b9ae3ea9c2d","subject":"Fix trove guest agent logrotate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3864fc262121881bd676f832c363b9ae3ea9c2d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:15:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:21:35.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","message":"Update TOX_CONSTRAINTS_FILE for stable/yoga\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nStory: #2010137\nTask: #45767\nChange-Id: I7333a75e8855501f8a89547b93e65157f9f8afb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37df560c2ba6c6ce0f775aec519e065833621125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37df560c2ba6c6ce0f775aec519e065833621125"}]},"branch":"refs/heads/stable/xena"},"eed7019ce1bf34dba30626ed1225c3275a063d75":{"kind":"REWORK","_number":2,"created":"2022-07-19 14:22:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/850370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/850370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/850370/2"}}},"commit":{"parents":[{"commit":"f3864fc262121881bd676f832c363b9ae3ea9c2d","subject":"Fix trove guest agent logrotate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3864fc262121881bd676f832c363b9ae3ea9c2d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:15:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:22:53.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","message":"Update TOX_CONSTRAINTS_FILE for stable/yoga\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nStory: #2010137\nTask: #45767\nChange-Id: I7333a75e8855501f8a89547b93e65157f9f8afb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eed7019ce1bf34dba30626ed1225c3275a063d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eed7019ce1bf34dba30626ed1225c3275a063d75"}]},"branch":"refs/heads/stable/xena"},"425f065d8896a7aff0ab82316bf4865cece9205c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-19 14:24:54.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/850370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/850370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/850370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/850370/3"}}},"commit":{"parents":[{"commit":"f3864fc262121881bd676f832c363b9ae3ea9c2d","subject":"Fix trove guest agent logrotate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3864fc262121881bd676f832c363b9ae3ea9c2d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:15:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-07-19 14:24:54.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","message":"Update TOX_CONSTRAINTS_FILE for stable/xena\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nStory: #2010137\nTask: #45767\nChange-Id: I7333a75e8855501f8a89547b93e65157f9f8afb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/425f065d8896a7aff0ab82316bf4865cece9205c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/425f065d8896a7aff0ab82316bf4865cece9205c"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
