)]}'
{"id":"openstack%2Ftrove~851332","triplet_id":"openstack%2Ftrove~master~Ia079b7b7f5d96858c7dfe4b3d103c0efd86ae68b","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28743":{"account":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"last_update":"2022-07-29 09:50:49.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-07-29 09:19:06.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ia079b7b7f5d96858c7dfe4b3d103c0efd86ae68b","subject":"Changed py39 from py27 in tox","status":"MERGED","created":"2022-07-28 04:17:52.000000000","updated":"2022-07-29 09:51:58.000000000","submitted":"2022-07-29 09:50:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851332","meta_rev_id":"ebf5e090ed59859d64a96aafd1019e0ccc0854d8","_number":851332,"virtual_id_number":851332,"owner":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 09:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-29 09:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-29 09:19:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 05:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-29 01:09:42.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"a5cc0eaa203b6bdb29d0a3dd8f3e5efa97533fd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-07-28 04:17:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c9bce66a95ee452560226b1e4ff3583e59bc966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 05:01:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345d83da3b644964a3551b769cd93e6d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1563fdd63c640bb92b5ed13fda78796 : FAILURE in 8m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e449143d8299484a943ff54d411279f5 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0b162493c4c4fb79acea34bc1a2e6d5 : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e2aa2103a154732ae81718a513a9562 : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eae23dff24ae485cb58fa5a3d5922a29 : RETRY_LIMIT in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99890149f155419a97a11426fe4ab5cb : SUCCESS in 7m 11s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2b1d06ff34b2469aaffa30c05c6658ef : FAILURE in 40m 06s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/85d63eb84cbd4d91b09d3334c24e9390 : FAILURE in 40m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3a5a198676b49beb518908f4bd8b235 : FAILURE in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba62584f77251f23c7bf5c1baf1510cf16890860","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-29 01:29:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f24133cb2237a8e09d4798ec4b51cee34a32e46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-07-29 01:34:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b903c21f35bd215429680f176f05e37292c13377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-07-29 01:37:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50947c08e34eb75950d18efe5b1de1979a9e3e81","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-07-29 01:38:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b705413b224f3c5566d076b33d5b9c8e448ff747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 02:10:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1d12e50cad7476ca727d8ced4efa9b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77a2161155714c51b271e4eff5dcc728 : FAILURE in 6m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2baec2365cb846ee80691a31da565673 : FAILURE in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23940a43ad0a4582a7a1f49e03d59a82 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f3612d2d4d240b6926f4f2481656147 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd50a428282844e9b7c07b52d79dd536 : RETRY_LIMIT in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4f6d6e643d486cbdfd77f7491be458 : SUCCESS in 5m 41s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/da112837113c4dd6893cb9c6e8f77eb6 : FAILURE in 31m 11s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/afe76e70d2524811ad05795a0604556e : FAILURE in 30m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7787d602b0e4469faf750c67702c8d04 : FAILURE in 26m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f8f61c6445cc4039dd6432f5047d3bf7f014fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-07-29 02:26:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"606996b9439783ae93ec3e3213c9d55eb1cd10fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 03:05:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f4238dce7dc40149542f0263cd19643\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19d8a7866aa24dcb8a2cf794ec72d66f : FAILURE in 7m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/832bdfa4080046ce887450e337db8a47 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0091c57fcefb4e0c8b7601fe5e65e405 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a90adf041bc54d388caf7a452ebcef97 : SUCCESS in 8m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89a9868859d94931aef0e7dfa7232430 : RETRY_LIMIT in 2m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b535d6a20789416b9f8bd1051890803d : SUCCESS in 6m 33s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2c1bdb58fd1242d0b5741d4a22402cd5 : FAILURE in 32m 34s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/cfcdaedeb7f94283a098599c0a764fb7 : FAILURE in 34m 14s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21000a2db72346ccbe1a01cfe20a6a31 : FAILURE in 32m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e187b2d9dc7790db814200c9e10e255e4ca7313c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-07-29 09:19:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"61322e37a8a15aa2c8a30e939f5c43dd871b7c41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 09:19:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"55f202b29878ef0559fcadca14a905813f316e9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 09:50:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e40313eacc584d97b11db8766d2c8876\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b37a3552bd948239e2139a5e37a0646 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47d509986f5445b3854536627c66c85f : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be78edec67e8462e887bd980841e4533 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3a433a5d674cb2bc451885de6b156a : SUCCESS in 5m 56s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/fea7ecbab8df496fb80f72bd8f5a92ba : FAILURE in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a71d66bd96b2ae80e03267008ff86d6dd762fa9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 09:50:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ebf5e090ed59859d64a96aafd1019e0ccc0854d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 09:51:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd04a9ecb3674fa1af48794a140663ea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8769294d670477d96df5ed03848a5f5 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"476c5f89c7c120a28a4b0b83a0ef81e7aeb02010","revisions":{"cfab358fb92af8d3dd14a8a13acbcb0980142169":{"kind":"REWORK","_number":1,"created":"2022-07-28 04:17:52.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/32/851332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/851332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/851332/1"}}},"commit":{"parents":[{"commit":"f4783ad4973481ae7136f19716ebbb8a6eba286c","subject":"Merge \"Fix import from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4783ad4973481ae7136f19716ebbb8a6eba286c"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-28 03:32:24.000000000","tz":480},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-28 04:16:53.000000000","tz":480},"subject":"Remove py27 env in tox for contributing doc","message":"Remove py27 env in tox for contributing doc\n\nRemove py27 env\n\nChange-Id: Ia079b7b7f5d96858c7dfe4b3d103c0efd86ae68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cfab358fb92af8d3dd14a8a13acbcb0980142169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cfab358fb92af8d3dd14a8a13acbcb0980142169"}]},"branch":"refs/heads/master"},"265617ab2121c60e5ad5d499d9e8183339569058":{"kind":"REWORK","_number":2,"created":"2022-07-29 01:34:01.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/32/851332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/851332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/851332/2"}}},"commit":{"parents":[{"commit":"f4783ad4973481ae7136f19716ebbb8a6eba286c","subject":"Merge \"Fix import from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4783ad4973481ae7136f19716ebbb8a6eba286c"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-28 03:32:24.000000000","tz":480},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-29 01:32:41.000000000","tz":480},"subject":"Changed py39 from py27 in tox","message":"Changed py39 from py27 in tox\n\nChanged py39 from py27 in tox for contributing doc\n\nChange-Id: Ia079b7b7f5d96858c7dfe4b3d103c0efd86ae68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/265617ab2121c60e5ad5d499d9e8183339569058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/265617ab2121c60e5ad5d499d9e8183339569058"}]},"branch":"refs/heads/master"},"2893d513571849570de06c3ac7dba9b516b8f023":{"kind":"REWORK","_number":3,"created":"2022-07-29 01:37:38.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/32/851332/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/851332/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/851332/3"}}},"commit":{"parents":[{"commit":"f4783ad4973481ae7136f19716ebbb8a6eba286c","subject":"Merge \"Fix import from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4783ad4973481ae7136f19716ebbb8a6eba286c"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-28 03:32:24.000000000","tz":480},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-29 01:37:17.000000000","tz":480},"subject":"Changed py39 from py27 in tox","message":"Changed py39 from py27 in tox\n\nChanged py39 from py27 in tox for contributing doc\n\nChange-Id: Ia079b7b7f5d96858c7dfe4b3d103c0efd86ae68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2893d513571849570de06c3ac7dba9b516b8f023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2893d513571849570de06c3ac7dba9b516b8f023"}]},"branch":"refs/heads/master"},"476c5f89c7c120a28a4b0b83a0ef81e7aeb02010":{"kind":"REWORK","_number":4,"created":"2022-07-29 02:26:52.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/32/851332/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/851332/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/851332/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/851332/4"}}},"commit":{"parents":[{"commit":"f4783ad4973481ae7136f19716ebbb8a6eba286c","subject":"Merge \"Fix import from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4783ad4973481ae7136f19716ebbb8a6eba286c"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-28 03:32:24.000000000","tz":480},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-29 02:26:28.000000000","tz":480},"subject":"Changed py39 from py27 in tox","message":"Changed py39 from py27 in tox\n\nChanged py39 from py27 in tox for contributing doc\n\nChange-Id: Ia079b7b7f5d96858c7dfe4b3d103c0efd86ae68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/476c5f89c7c120a28a4b0b83a0ef81e7aeb02010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/476c5f89c7c120a28a4b0b83a0ef81e7aeb02010"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
