)]}'
{"id":"openstack%2Ftrove~852857","triplet_id":"openstack%2Ftrove~master~I6dd126d49196af30bd5578c6aefa90fff9f1d5c1","project":"openstack/trove","branch":"master","topic":"drop-bionic-support","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-08-13 07:25:57.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-08-13 09:39:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6dd126d49196af30bd5578c6aefa90fff9f1d5c1","subject":"Drop bionic support","status":"MERGED","created":"2022-08-11 13:59:21.000000000","updated":"2022-08-13 09:41:31.000000000","submitted":"2022-08-13 09:39:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852857-drop-bionic-support","meta_rev_id":"434d5fb022b52d70d5df64b91c66705a37e98c36","_number":852857,"virtual_id_number":852857,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-13 09:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":2,"date":"2022-08-13 07:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":1,"date":"2022-08-13 07:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 14:04:59.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2022-08-11 14:44:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a615b30773c13c6f85f023758b668b8d175518bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-11 13:59:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebf4acad944da403e462074cf2d373e3e2f6f2b7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-11 14:04:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e28f580ab5235db0f4205143db1cc3580b6cad91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:44:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18c6497066824a998c111eb01be418cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fb58f1b39084ab9a41705c6126deb80 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76895d5103e94542988e9b735471ebf9 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/968990396a7f4fe298a92ebe0e8bdb1d : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2c68acc05bd48e7a58f739e7281b9d1 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e974ec627c0b41088f57c1cab3d51ca3 : RETRY_LIMIT in 2m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47df91f2a63486c8c727fe2037841d8 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f891518f2ae745b1b500bfd1234f8a6d : SUCCESS in 5m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e83ceb669042409f92ff5c986d451e27 : SUCCESS in 6m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f3ca1b50d42449ffba66383f3a5d63e8 : FAILURE in 40m 17s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/be23f752e1f24a4289f31474668ae54a : FAILURE in 9m 01s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/fedba359fbc8427c886ff2b2f0767d37 : FAILURE in 8m 58s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/87e324f3c3a7488e9a0e3b31ba7d07d0 : FAILURE in 42m 25s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18c9fce749c940dab09f17d150d09e1a : FAILURE in 35m 23s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d6189167d5254942a463f0c42c6bb424 : FAILURE in 26m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89e4fa1c91848e05d9ba40d7c03b98d6452e8bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-12 06:51:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"02ec6a84cf715749d85378eff3f32592b833090b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 09:03:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5fcf1d6db6b42ecab4fa9180f06455d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47da9690a6054efd9926c2b48650022d : FAILURE in 8m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba17791e32e94da4885b398176f19ae7 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1443a220d4b44718228c3a8869c11c3 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d47a89b945c84299b35942db3a3869df : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecf3be401ca242fc8b4e2af25852b894 : RETRY_LIMIT in 2m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08121112c1534e45aca6286ad27fc21f : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b514ad000e2b4de1bb8b422b99da81e4 : SUCCESS in 5m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d54f43ed3815404cb04479d0cc922894 : SUCCESS in 6m 41s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d2dc2089a02f48d0b4e7c1770f8994e6 : FAILURE in 2h 07m 28s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/b4c7bbba7f8547588837707e30d97657 : SUCCESS in 10m 19s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/ec49ed3f81ae46c093afd083b092ea0b : SUCCESS in 11m 44s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2fbebf6f29754d4ca3db5dcb19feab44 : FAILURE in 2h 11m 31s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7264967a903c48aa9931231b34ab5c52 : FAILURE in 58m 13s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1179d693ba7545a7ba720457ffe1ec89 : FAILURE in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c2167ebf5ada3b888f15cb9d260a4d6d52507e3","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-08-13 07:25:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"944be82af7ffd2d18287c19b72cc0f0faf2652b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 07:26:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"053642680c0356239b1682d3fbe2b0bf750883ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 09:39:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8610e0fd2fd472f86024a380f4530d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774b826c43ea4404ace01ae864564106 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61792c34e45d4215afef8beb6ce4c7c6 : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e8ac1f9d0954f34b90da690e7dea1c5 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a79fb5da4314cd0ad8002bf291ab006 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e635c79a18f04854b8de2e015eee9929 : SUCCESS in 4m 35s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b3aaa9ac55a04bc0803e918ca022f395 : FAILURE in 2h 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c4ecea2a2181c5b762e3f2e00d8cca7a10584e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 09:39:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"434d5fb022b52d70d5df64b91c66705a37e98c36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 09:41:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c42386633316462dabfeafa3db84f49f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f2c80ed19c4f7f8366aa21a4073d5d : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6266465623074b1789a119f723678c14 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ea76a145a2c39920088fa463c9c9e80d00e8338","revisions":{"871edb7c0a202fa71a8201f8b8c40621e1402a8d":{"kind":"REWORK","_number":1,"created":"2022-08-11 13:59:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/852857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/852857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/852857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/852857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/852857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/852857/1"}}},"commit":{"parents":[{"commit":"d126aacf9012a2dc572967ee75b4ccbf77fcc2ea","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d126aacf9012a2dc572967ee75b4ccbf77fcc2ea"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-11 13:53:17.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-11 13:53:17.000000000","tz":480},"subject":"Drop bionic support","message":"Drop bionic support\n\ndrop the support of bionic.\nref:\nhttps://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I6dd126d49196af30bd5578c6aefa90fff9f1d5c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/871edb7c0a202fa71a8201f8b8c40621e1402a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/871edb7c0a202fa71a8201f8b8c40621e1402a8d"}]},"branch":"refs/heads/master"},"8ea76a145a2c39920088fa463c9c9e80d00e8338":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-12 06:51:00.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/852857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/852857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/852857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/852857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/852857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/852857/2"}}},"commit":{"parents":[{"commit":"136bd3f1c9b9e9083c51c532f297244ebf6ec03a","subject":"Merge \"Increase guest image size to 5G\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/136bd3f1c9b9e9083c51c532f297244ebf6ec03a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-11 13:53:17.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-12 06:51:00.000000000","tz":0},"subject":"Drop bionic support","message":"Drop bionic support\n\ndrop the support of bionic.\nref:\nhttps://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I6dd126d49196af30bd5578c6aefa90fff9f1d5c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8ea76a145a2c39920088fa463c9c9e80d00e8338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8ea76a145a2c39920088fa463c9c9e80d00e8338"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}}]}],"submit_requirements":[]}
