)]}'
{"id":"openstack%2Ftrove~854356","triplet_id":"openstack%2Ftrove~master~Ibbd6ecb8db25aa3a443d3f526f4cf6d418eb26f3","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2022-08-25 11:26:23.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-08-25 11:49:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbd6ecb8db25aa3a443d3f526f4cf6d418eb26f3","subject":"Rename app.wsgi to app_wsgi.py","status":"MERGED","created":"2022-08-24 09:24:28.000000000","updated":"2022-08-30 15:31:59.000000000","submitted":"2022-08-25 11:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854356","meta_rev_id":"d3920e344f3f82355c0241ce69b7addca50b8a40","_number":854356,"virtual_id_number":854356,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-25 11:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":2,"date":"2022-08-25 11:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":1,"date":"2022-08-25 11:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 10:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-25 06:42:37.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"1d6d811ac0da672c59c8bcf8cb263a603db0c253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-24 09:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68eeeecfbdf5d4e37f081a072fa52554692793ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 10:06:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ca15fad447444b2b76d6ddb69112c08\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/038ed52045434cadaf567196bba55fb1 : SUCCESS in 8m 34s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56c85de1f8344bb5966c36f433d55116 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d5dbcef99ad467597d2778c55523baa : FAILURE in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b63ed652d6249b08b8188bd43e07e8b : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0332c2648604582ac336cc397144357 : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e60befd5b3094c7881e5a73c584e9733 : SUCCESS in 9m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4d30f48e1ef4c0fb932452352f99337 : RETRY_LIMIT in 2m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5dd80fe2d9f4644b4fa7bbf7b015d30 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60a8b8486a214ddba994ce6601a04a95 : SUCCESS in 5m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/162b8865fcf843ca957a2a5f6146a4f9 : SUCCESS in 6m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2017fc6e4a1546a3ac49b49e8fdc4a80 : FAILURE in 22m 01s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0fba2ecbbdee4f3da14ed946081b8d7b : FAILURE in 21m 35s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83aa2ce7ab6a4e05b0e839f9cd73743b : FAILURE in 35m 07s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/292a6e77ed0749239ca598c71ec32df2 : FAILURE in 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29baae031e5508f1a93e144ca6b7e1583b4fd14f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 03:10:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2980e6f6636ce424f08edfbc5f8812ad5a540ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 03:50:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6bf4d8dc05f4fb3a155eaa8e527870e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3c1f68bf10ec44d39e44d28c97c144ad : SUCCESS in 6m 17s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/371efdf068334b14b852b60f9c1b9fcd : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f82b9e37b644b67acd13338d479832f : FAILURE in 8m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1bc4a1eafb4e42a2f76fc31887f33b : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b309e1ee2914b09acf5f9de716b55fb : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98db4d516a5a4ec28a3590fb781c9450 : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3cc14b9fd3e4b85b5a04c994b92263e : RETRY_LIMIT in 2m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1abf31968cb4d92bbc4b448b8c2f4d7 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f0410e1a7a340eea68c7e9654866cfc : SUCCESS in 5m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd6a901d21fa47f4935f8e8ed23b9d2f : SUCCESS in 6m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1a8f959b27524c3fbdda6cf795d8b011 : FAILURE in 24m 20s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e5b3671624c24a16b2de84f8dc92347c : FAILURE in 28m 16s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3fc7c877c1f40f4bafc2b5201a0f21e : FAILURE in 34m 47s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/26c23eac43644b2e8b53bde40091aaea : FAILURE in 26m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00072f97a093e16d27e5beb5008e922f4207a49f","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-08-25 11:26:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5cd24526fb56da85de522330a62db58aa6549b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:26:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7bdb5ff78f463e3c65f7cd45fbe3c3712e0b77d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:49:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8638a3eae7346a29b66e7be9c6fcb33\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ae716a24d4df4975a24234b60e75c461 : SUCCESS in 5m 38s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5af6222e6ac4d5f82998e843a96d188 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28505f36f394313a1db5e9fec4a93c6 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad5300bb8c9c40ccacfe98423a925327 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c687ed8d5b7342aaaacad9ef9a47e7e8 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fac5b6c7a24bf5855ac530b8e37363 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/042852882a5f4164826430318d283d9e : SUCCESS in 5m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b36916fd3bd84136aef7b77e9b954138 : FAILURE in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c50bb34ef08abcf3e434750b61034057fd7c1288","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:49:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0733ed64b195c8590dbf58351ab9f898aa3d29b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:50:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03cf630a741f4700a97ff484f9703d0d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/10f08b54898b43e4b6cced34874ab760 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d562381ff03b4999aa3fd5acf19b3c2d : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d059efd92e4545dea2d0424219a567f2 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"d3920e344f3f82355c0241ce69b7addca50b8a40","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-08-30 15:31:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","revisions":{"715ea5884f556a7b8429788c3943e01cb46d5a34":{"kind":"REWORK","_number":1,"created":"2022-08-24 09:24:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/56/854356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/854356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/854356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/854356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/854356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/854356/1"}}},"commit":{"parents":[{"commit":"528741b449a843d118c527d160dedebb523fd49d","subject":"Merge \"Fix the missing of guest-agent.conf in guest vm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/528741b449a843d118c527d160dedebb523fd49d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 09:17:42.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 09:17:42.000000000","tz":480},"subject":"Rename app.wsgi to app_wsgi.py","message":"Rename app.wsgi to app_wsgi.py\n\n1. app.wsgi is a python script and should be end with py extension.\n2. Generate trove-wsgi script which is needed by kolla to support\nhttps and ipv6.\n3. Ignore build directory in git\n4. update docs\n\nChange-Id: Ibbd6ecb8db25aa3a443d3f526f4cf6d418eb26f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/715ea5884f556a7b8429788c3943e01cb46d5a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/715ea5884f556a7b8429788c3943e01cb46d5a34"}]},"branch":"refs/heads/master"},"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09":{"kind":"REWORK","_number":2,"created":"2022-08-25 03:10:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/56/854356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/854356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/854356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/854356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/854356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/854356/2"}}},"commit":{"parents":[{"commit":"528741b449a843d118c527d160dedebb523fd49d","subject":"Merge \"Fix the missing of guest-agent.conf in guest vm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/528741b449a843d118c527d160dedebb523fd49d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 09:17:42.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-25 03:10:33.000000000","tz":480},"subject":"Rename app.wsgi to app_wsgi.py","message":"Rename app.wsgi to app_wsgi.py\n\n1. app.wsgi is a python script and should be end with py extension.\n2. Generate trove-wsgi script which is needed by kolla to support\nhttps and ipv6.\n3. Ignore build directory in git\n4. update docs\n\nChange-Id: Ibbd6ecb8db25aa3a443d3f526f4cf6d418eb26f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}}]}],"submit_requirements":[]}
