)]}'
{"id":"openstack%2Ftrove~854940","triplet_id":"openstack%2Ftrove~master~Ic346ada2bdf318457cfe5cc6c4332fc55bed5090","project":"openstack/trove","branch":"master","topic":"ignore","hashtags":[],"change_id":"Ic346ada2bdf318457cfe5cc6c4332fc55bed5090","subject":"Ignore swo/pyo files in git","status":"NEW","created":"2022-08-29 05:37:00.000000000","updated":"2022-08-29 06:03:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"932e05846bd8e5007be1f801ffa24705c696310c","_number":854940,"virtual_id_number":854940,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-29 06:03:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 06:03:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"261d8069e9083e0d9872cb538141c8bfa23e034c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-08-29 05:37:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"932e05846bd8e5007be1f801ffa24705c696310c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 06:03:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8469339ff0af4a82b51f1c17300e6764\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af4d95264f1f4087a8dc4f869ce2c2b8 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f25ed29456400186b601b9f4420a15 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe67f8bbde5d4961be0aabf8551f3bd8 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfb9b36a897346148cdd0f1a22e9bb66 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a94dc80d63e24bffb8a468b68a793c1f : RETRY_LIMIT in 2m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087fa41878da442eb48c1eb6737ab797 : SUCCESS in 6m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c129ef12b99481f8b9d18e55806b242 : SUCCESS in 8m 31s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/14b06563364f49cfba6ce91b7c34a275 : FAILURE in 24m 42s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c50ad5a1fb054e9ca20b70c23b3ddd79 : FAILURE in 25m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddacb64fc17d45d3917000113c5e606c : FAILURE in 25m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e06ab46e9ae14c41a52aa83b8029191f : FAILURE in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6efb7ec39595ff1e729c302d85d10c62bbe59783","revisions":{"6efb7ec39595ff1e729c302d85d10c62bbe59783":{"kind":"REWORK","_number":1,"created":"2022-08-29 05:37:00.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/40/854940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/854940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/854940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/854940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/854940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/854940/1"}}},"commit":{"parents":[{"commit":"44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09","subject":"Rename app.wsgi to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c3cdef5c31bdcdc8e4961ec9d8ce1647aa8c09"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2022-08-29 05:34:52.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2022-08-29 05:34:52.000000000","tz":480},"subject":"Ignore swo/pyo files in git","message":"Ignore swo/pyo files in git\n\nTrivial Fix\n\nChange-Id: Ic346ada2bdf318457cfe5cc6c4332fc55bed5090\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6efb7ec39595ff1e729c302d85d10c62bbe59783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6efb7ec39595ff1e729c302d85d10c62bbe59783"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
