)]}'
{"id":"openstack%2Ftrove~857584","triplet_id":"openstack%2Ftrove~master~I814cbc2eecb94c99efcc51f7c46306da10c5fe50","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-09-14 12:25:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I814cbc2eecb94c99efcc51f7c46306da10c5fe50","subject":"[CI] Fix CI failed","status":"ABANDONED","created":"2022-09-14 08:10:59.000000000","updated":"2022-09-14 12:25:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c15c013c2c3747aa015a859ace9014a1f79ec948","_number":857584,"virtual_id_number":857584,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-14 08:52:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 08:52:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfe1c7b7bc37fee4efc4acdba0d8a0f8cb89db51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-14 08:10:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b966b8a86f5270498321145dea86954466b2af04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-14 08:13:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4352d576b497d936f1f87edcde16f867735b505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 08:52:49.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b376b4b2ba450bbf85e90c3e112f20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87da4dec216c4c27ac8c7810d24f04fb : FAILURE in 7m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9947d367ba274d9bb3f41eb1e393225a : FAILURE in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e341ce3d11aa471a91ae4b91a012039c : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba34af50a9b3498984ffc24610909ebb : SUCCESS in 7m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5588282598804377998c794cc59a968f : RETRY_LIMIT in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8c25c3d4c44c248e5f0825e6e32bce : SUCCESS in 7m 59s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bfcd776181d74e09b5e027217f1e15fa : FAILURE in 31m 19s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a0a85fa315eb427c9fa9aa9e90c87b95 : FAILURE in 32m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b1b3bb9299d4253ba26211d3b24b410 : FAILURE in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c15c013c2c3747aa015a859ace9014a1f79ec948","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-14 12:25:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05afdd00a0f11a2cddbf83b195a8108f5b72b6a9","revisions":{"622110aa92b5a9f877bd12bde0b6e0088b348d6d":{"kind":"REWORK","_number":1,"created":"2022-09-14 08:10:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/857584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/857584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/857584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/857584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/857584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/857584/1"}}},"commit":{"parents":[{"commit":"6e4dd51c02f042efcfd6fb89575f100d44ac0a60","subject":"Rename api_wsgi.py to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e4dd51c02f042efcfd6fb89575f100d44ac0a60"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-14 08:08:45.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-14 08:08:45.000000000","tz":480},"subject":"[CI] Fix CI failed","message":"[CI] Fix CI failed\n\nAfter oslo.db released, our Ci tests start to failed.\nMore:\nhttps://lists.openstack.org/pipermail/openstack-discuss/2022-September/030317.html\n\nChange-Id: I814cbc2eecb94c99efcc51f7c46306da10c5fe50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/622110aa92b5a9f877bd12bde0b6e0088b348d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/622110aa92b5a9f877bd12bde0b6e0088b348d6d"}]},"branch":"refs/heads/master"},"05afdd00a0f11a2cddbf83b195a8108f5b72b6a9":{"kind":"REWORK","_number":2,"created":"2022-09-14 08:13:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/857584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/857584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/857584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/857584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/857584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/857584/2"}}},"commit":{"parents":[{"commit":"6e4dd51c02f042efcfd6fb89575f100d44ac0a60","subject":"Rename api_wsgi.py to app_wsgi.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e4dd51c02f042efcfd6fb89575f100d44ac0a60"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-14 08:08:45.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-14 08:13:06.000000000","tz":480},"subject":"[CI] Fix CI failed","message":"[CI] Fix CI failed\n\nAfter oslo.db released, our Ci tests start to failed.\nMore:\nhttps://lists.openstack.org/pipermail/openstack-discuss/2022-September/030317.html\n\nChange-Id: I814cbc2eecb94c99efcc51f7c46306da10c5fe50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05afdd00a0f11a2cddbf83b195a8108f5b72b6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05afdd00a0f11a2cddbf83b195a8108f5b72b6a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
