)]}'
{"id":"openstack%2Ftrove~861781","triplet_id":"openstack%2Ftrove~master~I13405e8bc93586441da28898e99dce936fc6d68a","project":"openstack/trove","branch":"master","topic":"disable-shell","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-12-18 08:19:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28691\u003e replied on the change","reason_account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-12-28 07:09:52.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-28 06:34:43.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I13405e8bc93586441da28898e99dce936fc6d68a","subject":"Disables shell\u003dTrue in backup and restore","status":"MERGED","created":"2022-10-18 17:14:42.000000000","updated":"2023-12-28 07:57:02.000000000","submitted":"2023-12-28 07:09:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":51,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861781-disable-shell","meta_rev_id":"40dd8e9bbd6fee1f96242519159f3e938f6eab9b","_number":861781,"virtual_id_number":861781,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-28 07:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-28 06:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-28 06:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 18:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 14:04:54.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-06-21 01:44:07.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-11-07 09:38:24.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"c4178ce3d4bb81144c3368b3ef093ebb16ebe166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-10-18 17:14:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d9929fd103ea96280b5a4ed2d1af01a25c39b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 18:28:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/109ccc9473e440caa296daf6c7a93758\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d1174b8a2f4990ab2f7a2ff5d173a9 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdcd307eca1842bfa79e7143b97f1e2a : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85ad0e7a7d334406a3dcda9fae800b53 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a7dd366cc7049bfb401b8abe1866581 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e482fae0ccd4a6794c86f8cd37a1c09 : SUCCESS in 6m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac07f017d6914a87be1b0b71a935c195 : SUCCESS in 4m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9fb0c07fc3804b3086455c7da7dfc584 : SUCCESS in 1h 12m 43s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/74f652af335742c1a312ef345f4972f2 : SUCCESS in 1h 11m 27s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/737e0f6622374cfd8d74824a5ffa4f7e : FAILURE in 31m 49s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/13404c29741844a3a9d263ab08c4a758 : FAILURE in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f9fd5654cf3d45d9c944114fe44a13e16cebe2a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-10-19 14:04:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d78a2eecd59290c9e2781d6231a896cb6b210a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-10-23 08:35:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdb93e10a0a8cf48044bc5c8f7cfca7798e4fdb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-10-23 08:48:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"11b042c7a1d739b60e5232eebd4f8b2840678f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-10-23 08:49:04.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9f3ba77bc2667eb70d84c653acbfba4162394f8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 09:09:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7ae002f72348fc828e6cbb91820543\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa9a3a8438fd4f0fb79566ad1a15cdb0 : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66b5867dbaeb471da2b5b709635cf4e7 : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"745faa47157e02a8f16fe3b5203a890f05ec742b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 10:00:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/351f76b491b54a27bbc03ba563101662\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3dd8991f77e455e97797a05ab38662b : FAILURE in 5m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf7a891f0524b4f8a3fda29c87a08be : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4802f92153b42b4ab506742251ff3fd : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df4f96d3b4214482b6a38506cb84b02e : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624f265701734e7298d84a3b621676ca : SUCCESS in 4m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4bcc555a14ec4b22af17f127f1083fa3 : SUCCESS in 5m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a077ef1d8b54457eb6a8c0c0c5cdb36d : SUCCESS in 1h 10m 50s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/76bc6bb8a9d64504b58f55616fbb45cc : SUCCESS in 1h 09m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4d741bfa1754e2aa9d9250ef2da404f : FAILURE in 26m 34s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0f864f845de04972a1ab9dec1db330fc : FAILURE in 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4f547c7c0958f5ef50efcb316a5dc42424b35e91","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-21 01:44:07.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c5b91f78bf727526d1703d7dd4b1f49a972e1c8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-21 01:44:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c384789e1006bea6490611395a9839d32e84d27","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-07 09:38:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3fc10c205bdafd4ba27ff644fb334b23a87af135","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-11-08 15:31:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"910e8c8bce0359f3fa3ed802a1cbd8c076622941","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-09 01:37:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"15973e1670abe0b68fbe22cf373d793c82ebf082","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-09 02:54:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"14b6d70d18c7343ed269ff90487ef80511ae807e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-11-27 16:30:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a71ce9a3ec4dcb0c5a850688a9bb9e29453bc033","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:37:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b279b3aa974f309d4b1810e1668b83\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bd4b21f9b4d4869aaa2b56596a3df8c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1ba2afb3e5f4176a70c028a222cfb5e : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/811bfc7e8e6d45d584f3dd4085c7cb5d : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff4b04c8fc25013742d6c3856035fc0dee923e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 18:50:39.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f5d67f756f5481cab6a6ea7ac04b3fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26c9fa62ae844b9c94f35cd91155bcfe : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4af34facac418e9d92cfce36b491e1 : FAILURE in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/164bd0ab558141db816ec5bda67dc92e : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31d041ed3e274a51a883865d8debc7f5 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4357d538f85d49de9a65672edb30306b : SUCCESS in 5m 36s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d6aa752141cd4543a178fb7554d23eb6 : FAILURE in 2h 17m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3a247a513c0c47538170d5a9a189cace : FAILURE in 1h 53m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9c8e417be0084db093e69e81c8a0ef37 : FAILURE in 2h 18m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6da0dc66656426985c599cc23347587 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":5},{"id":"257888aea82674633ac3fb6af77f414e50188d46","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-29 03:24:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45cc440cee977f4445a672a5f6a713f88e1064ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 09:43:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"428d823375ced0c1420c0ae7c03cdbf5f9143a9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 09:46:47.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"87166edf676911adaead45193e77568847e1fca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 10:02:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dab67cd2b644182b525cf8440a03032\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f33e1ebc51b1411996223a8640b36b48 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/feadebcd39b14d0ca7ea8a473e11e6c3 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a8890dfbd6b42d3965ecf0d63cc2f91 : SUCCESS in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"400ba4b7e92e6b61a53ad5aca42a93e7a3057245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 10:56:49.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4045450cf31e4702ab3ad3196558bfaf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/50c10c180d5c48a19ad62a10012d23ee : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a4c6c631e2a43828ff676a52de66c8d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22977466e58247219fdeb633744f6897 : FAILURE in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73aa7b03da3e4e6c8fdeda0f75176cb1 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0b868c072a64d3485b49a9bfb087065 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07733b0302e49d391fa303b9f5955f9 : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2cfd142010ca491d92bfa431b2bc7afb : FAILURE in 59m 05s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ac3d605d3d1e4278b93003c1e8a747d5 : FAILURE in 54m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/fc3954eeb07543ca882c33a33845e87f : FAILURE in 58m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de87d77c748743fdbcfbd5433faabcdb : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":7},{"id":"ded5d91389d98ed7f2655838355b389e6323544b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 09:07:21.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8b604b2fa43d4d0b5784b69c634173451ba78220","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 09:22:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3851309074474812b52bd9acee095e04\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be29f0b4d33f4dfeb4714092a06e5b59 : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8078cb49a99e447ea66f73cb2221006f : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba521cdd46244ac1bd494dd38d675602 : SUCCESS in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7ad6b81db03f4b4034981cd66d8ba76a24a884dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 11:45:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60fb5f252ca246b593c17186a9d4e803\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ba8db7417634e8e97736ea22672bb2e : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1791586d2564aa0b767d27a8d3f843d : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9b6e3a8c0c429cb352f51499f73cd4 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d482bc65564051a5ecff586e9312a1 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59da1c735b164b2ab25c0b3de8fb7b4b : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23e6aae15634afabeab65849583efae : SUCCESS in 7m 35s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3a836c155e764925989f0959e770a128 : FAILURE in 1h 25m 24s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/39f240b354734123bcb8c99c637c6eff : FAILURE in 2h 25m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/10776b233a7f4421b5d61a6836d1cb60 : FAILURE in 2h 23m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23824fee75e94e4b913b9c964cb38c53 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":8},{"id":"3f3f754be4345d8413e39666ca215d1b22a71fe2","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-10 11:51:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"72dd1a5575582071c0a54228967d43579be59e54","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 01:20:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ca8bfd6e9cac64bd14318fc9509326143a3dfde8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 06:58:03.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5b835a1053ce1e867ea108c8f19d0da741dfcbc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 07:05:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae5557aad5e4509a6d34c06810894c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a60da72dacda41edad5e32423e9df467 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbd6ee66cc6d4e96b2a2e3f5631e7f93 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02e75e5b08874c198e7e60ad0543cf3c : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"897c2504a138569fafe86dcfaeb90fc79ff4d396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 08:58:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a4538fcdeb4ffaafd3220b7b6a36ff\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/30ecc1c3f8af4e1fb849f0ed684d581e : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c10703c344b0490e9b7e7c13c7a41e5c : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2ce629e9c844b590f29029b4ca3066 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6c816b6f5904049a88fc68fc94f4235 : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50298d904cfb4ec1a712b5c54ac15461 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7249a5c365fa403880d4fe83d495f677 : SUCCESS in 5m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d971a0ede1f141cc8d26da0e731708da : FAILURE in 1h 47m 32s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ea9ec07189244e1c9bf3cd6809a4009f : FAILURE in 1h 48m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/008d589ea1764673a68bc8bef63ef049 : FAILURE in 1h 58m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d0fb5214d1b4708816e40e292166fef : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":9},{"id":"058c7cfcd4a02f5c185833d6ce25a9b903f7ee6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 06:54:07.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6d19fb855f50e1eb634366f8100ce97990b8410c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 07:00:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128fc695a75b475a91c56fe32769e577\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87c0dc4318b84664834d0f99c0429abc : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bbb5b4dbaea41d9986aa417072742c1 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7f1ec210914043b8ba10b974b9a140d5 : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c56d847dbb396b561e0b553efa54ceede59088de","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 07:01:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5568aa3ac0c8a168a4f7668a62ca447c4390f40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 07:16:44.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8b973360699c9f374f59560cbf6e75df4c47b155","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 07:30:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205f043a498c482a85bf5358344f689a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0f65e528b8f49afb9047f92ce22425b : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c00d8c9cd384722bae53b22ad786043 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/321bbe434f1d44459775363ccc52e50b : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7f6b3ca41717a86d63156f88797559acf4d73765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 08:56:53.000000000","message":"Patch Set 11: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3948ba46d6484eb980201148c79cb74d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/84493ab3dd13443fbda85df97fb00c5e : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d35d66f7604a139d0de1935a0a2013 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a619f382d15f4faa8ecdf46fe35b5070 : FAILURE in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f96fa76de24f11b73cab6c6c5f23d7 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8a4d007cb31433d90566ce120056845 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb85a9c52104541a6e8edf85568b88d : SUCCESS in 4m 47s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9a4d14ca6c0b467b967560a964544283 : SUCCESS in 1h 39m 02s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bb65a6c791904df9b39ef714284ae017 : SUCCESS in 1h 30m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/aacde2d2485c464199f22147857f41f8 : FAILURE in 1h 29m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc7f45206b304fa2a986c94d02be21dd : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":11},{"id":"d2e88942138f6a76749836dd38cc2ca86420d35e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-17 15:11:15.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"77670f6c0e68d72a0172b52f05307cada45f668a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 15:23:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88aae71d8e243ccac2f3977a329c0fd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37f43ae520ee44ce84b7553c6ac04a09 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b09b11375b8430782bf927eb140127e : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0e867171775421bb89e64fa76bac891 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b66d825598c08b941d29e674f0e13e67a0b9bc1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 16:56:36.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/649e836ab44c40ab99c3787879106f04\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0ec38b5f97c414d84efe722b26aaa73 : SUCCESS in 4m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e7eb2e82d72425ba68dd8c6d2dbaac5 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/febd8b9169824d209855e39f74dc8bf9 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79825e17928e41b7ab6a86faed404b9f : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdc748edf5d84846928839b4da288e49 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d282fa012958447882ee062789ea4018 : SUCCESS in 6m 19s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6afeca22f1554188a9871e2c6255e0f7 : SUCCESS in 1h 42m 53s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/30f53e1f8cc64d7bb95b0dd480859ae9 : SUCCESS in 1h 38m 25s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a80ebe012849478f9a07f2d41dbfc585 : SUCCESS in 1h 33m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c288500b729c43cda572bfeb6b2fa2ed : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":12},{"id":"27f4b099df20f85d6c274900053a0ef680b0c55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-18 01:25:58.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"bbf6b0a7c1c4d78d9b4f1324118b41106b456fec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 01:42:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd2ce297d494872a50d157a7c9592ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9b05a4eedb546dea946efaa5be07de9 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/483208cbe3e54062b5bd34bed7eb44ea : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3555ef3497e74ad285f5467a0c49c5bc : SUCCESS in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2fc397bbbdb5f97dbd05bba051564ffb24d77ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 03:05:41.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23ca23b71aea425c81dd7e6dc20a6b01\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/64769c5c84f2482fb542740771f3ecac : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/287769e9c3694c349823ca257136e34b : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54484e6cc8d4c2a8e069bdfc3d20cf3 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcc00be2e35e4b11856332f194d68146 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fc4283842764e8dad6554b743068a27 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7e2bfe005a42b7b7cbd4f2902f6ea2 : SUCCESS in 5m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/84734a3563824b8f8d94b6c2a0897397 : SUCCESS in 1h 37m 32s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9ac42a20c2204aedb48dd6ae83653ba4 : SUCCESS in 1h 29m 04s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/50892874f4fe461cb4b90331746564e8 : SUCCESS in 1h 28m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5757e0407d242d3873cbd4c7347bd3a : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":13},{"id":"18cea56df8441a2a597ec7b79a0a098564a804a5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-18 04:01:05.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c7a26bfb04e1c33d841496f0c4d5bf6b68650262","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-18 08:19:50.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0fcb6ca9daac7b9c20fe27a513c9cf2731b4ad1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-18 09:26:48.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"d6a065f0e1c01efc3bb13bb491a21efa92d41ab3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 09:43:06.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5522b740d8594bddbc8b96043b4e0b49\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47038d68946b47f0ba91b17f77e9e8f2 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c88059b8567485f875f75c38133695d : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d05675ce27a14627a594b13f5984a7e6 : SUCCESS in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"eabad639b5425b85e1147b19d1ce1dde7619a255","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-18 10:29:17.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"cd6d2b9a14a7a8c527dec37703b5e84fbecc3cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 11:53:42.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71f783e20059491c944bc1e36d2ba921\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d44fb07abb9a4b7585f11011285e96d2 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c88681aafd4450beca98b1c4887768 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eee5c2c06df4bcca6bc8b3e6fe923ad : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99fb1f498f884e9b9a6234a8323e6bff : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab827a1a276d4dac80b0d5523681d73a : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c6aed2d21049669dff553b5410dde6 : SUCCESS in 5m 28s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3f9faa02d9854e38abf614f3b6669133 : SUCCESS in 2h 25m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d7f34b5d46c4412f9974b118ab37b9a1 : SUCCESS in 2h 07m 35s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1636f089f1bf4140afe9b3448ede913f : SUCCESS in 1h 58m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4102b3c53da421babfc05982b380d4a : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":14},{"id":"e8019589a8e0e58894dd07d79d817ab6d9899d1a","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-19 08:44:09.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"677772d4f94901e8917b5bcb416d94141e41db40","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-20 02:49:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0ea3dcab82364fbb351049006998100d5b7a2a6b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-20 03:29:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0d5cede3297a3a516186abd5d500893172acc02a","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-20 05:43:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e5c6ab9fef9be969d7147f59ec342805f1ea94f0","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-20 06:17:17.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"fb4434c83ef8509ca55acd5d77e5d39da11db3fe","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-20 08:22:25.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"324d9520f4882ffa28b2ca2da5b44138530d0b0a","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-21 02:30:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2d8b577a8fbf2f23352aa434d6a2f304e72f6160","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-21 02:30:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e4931d6ed04c2a8c0bb099ece26d5b992d6f6e49","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-21 06:25:27.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"98432f234f3a09ffd3e82fc565b914a17bdf76a9","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-21 11:28:25.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"31270ddf0c53893afb2417e73dcb89caee6734c0","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-21 12:07:46.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0cf966bee4a6dc1710fc3b9d1d97446d6e9310a9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-22 01:46:04.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d4dbd16e5fffaad45b3e84f571d166cd6f24447d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-25 02:03:52.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"ccf871a1dd19f7946d0ff4eea2a53700191f73ec","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-25 02:17:35.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1f641a733d51123d61fe96ae137e6ec2f9f49c2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 02:31:02.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0ff3bb1d9e403cbd4a239acd913c73\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d700197d433646c780c5c082e14105b0 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31f1c78e4f484f668cb49a2a7531cde5 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab0e10ccf1c6490b9523052dc46e2254 : SUCCESS in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e99e4ee85d6f5054c87aaa7fae0c443bea637870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 04:09:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4a602944a94fbfb489a84eaceaba43\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a74637e9e3004c0e999f6116cab3d9b3 : SUCCESS in 5m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/980f634b6b8645af856fc24bee8dab68 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a23a2159fe974d1b92b44e44b446b95e : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef09c28671804c608233c619d0deeba7 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/784365a1b41e42158585bc2cdc1499bf : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6204c349655341cd97a06ae7e12a55cc : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01867d30a03545c3a9a8a67cbac13920 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec0877d166b45f4bd74eec4f1c5889f : SUCCESS in 6m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4b4ea3a7a726420e980031add4e0c274 : SUCCESS in 46m 46s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4df7b11a5d15479cbe2b8d45a54c597f : SUCCESS in 1h 47m 15s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/38937472abed45ab8b3bf2ac19177184 : SUCCESS in 1h 18m 59s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3f6fdc2a00224517b9f228f20ed518e3 : SUCCESS in 43m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/527748352d714a14b0fbc8988769a295 : SUCCESS in 1h 25m 29s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6dbe8a7debe0487291d972ad38c77599 : SUCCESS in 1h 30m 25s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1db880f9f195474f9d59c19f94472964 : SUCCESS in 1h 06m 42s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/251c4693825443169cbedcb112669bac : SUCCESS in 1h 25m 50s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f1928e9155f6493ca5dd0787deb600c8 : SUCCESS in 1h 24m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1ca5315599b42ec8f4fd4b7a1d8ab86 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":15},{"id":"75475d237c3837fbd1544e116b27019189c319a8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-25 14:35:33.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"9fc6b905e189545575c44fa9d9b5d735460edf6c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-25 14:35:39.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"6a462b6a64988b1b78be282f3e2719b0a030b092","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-25 15:12:52.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5026b18c5c6102903ca93e24cdb23754db35eeb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-26 02:17:43.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"103e9ec57ba01dd9b2dac0aefd4cdf25e57b2270","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 02:38:32.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a28ffa2e23954f9a96084757e8d8f4b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f66d6c1d3ac4597a413f53d2f18c69b : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e41f7a6e27f4131864a22c3ba2e1c35 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76678108f4994e13a4d24d3031593b06 : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bd804bc9a82289bf8e44228d4174bf6df2bdbe3c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-26 03:27:00.000000000","message":"Patch Set 16: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"7f5489cc443cc7ddc63bf20f338a532538188dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 04:06:50.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c38aad215774767b7399a97987f2897\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0857a5ce265a4bb892126d51e0c61c92 : SUCCESS in 6m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/42885dfa37f8455898a13bad2d92db1f : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d40bfed4a82415b98e45bafc8595490 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06faf24e5ef04c548b5b5a0ef43c2caa : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2164badb39a349459cb3063b1377dffb : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca9e5ba37a54477895ddc201eb7cd9af : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd538039bb7435da7c6910752df7a70 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/329fdd717942434fb5ba25c77f9f3500 : SUCCESS in 5m 43s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a89910c247e84ef38988772b14f5774f : FAILURE in 1h 31m 26s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7116188a11264a05978e2e23fc634e60 : SUCCESS in 1h 44m 21s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6662659ea59247a9874034c5fde41d8f : SUCCESS in 1h 25m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c27641b8fd414b21895cc98d9b16e1ce : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":16},{"id":"e683b404968f7b3866fcb90ba175b6ee5219c72d","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-26 08:08:06.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b8bea2571fc258bd33a6505bc23e30e59f735339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-26 09:43:10.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"5e09afc3d1ef59179df722e2ae47e4f4a7d844f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 09:50:33.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f10b38ee0045a4ac138c9ef36b6853\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/916f254ac8f1477080f57c7a89653e34 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1532dbebe104b6fa953598433425f97 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d4a9bfc6dcfb4f2c9492039ccb19aa57 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4edd8ee56816da62e38febf4808d1588a0a0bd53","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-26 10:24:19.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6e8fcd517a29976b384babe1302b85b35fcc22f8","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-26 10:25:19.000000000","message":"Patch Set 17:\n\n(5 comments)","accounts_in_message":[],"_revision_number":17},{"id":"85cc4a16ad750e3c7895454512c01ec316d92296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 11:39:31.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d7bee741b64ea0ae997bee98d3e920\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c4dffaec47bb4974aa50ec2814070711 : SUCCESS in 4m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/39f3ddb6833d41b5b0859fb2efdfdcdc : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be0e0a963dc1496f9a18e2e3bed37080 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8cd670c0e4542d6a29358aa4c525376 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28167f7901254ab68ddbb0d9e2c59452 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a8947b6e43a498f9a9259c66cbb34d4 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49acf5659e14459a23b5667e308810e : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2848deb3f9354d75a6eecf23138b8530 : SUCCESS in 6m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/0850909bb6ec4566a7234c1bbbae6135 : SUCCESS in 1h 54m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4895d5833d30405ebc226c25965e1ee0 : SUCCESS in 1h 46m 54s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/3304d09b323d433c9dfb3792cdd1a482 : SUCCESS in 1h 54m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50dc94969c3c41189ac3653a964106de : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":17},{"id":"7ec2565d85dd636b12a225a202cdbf2ce1d1dd9d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-27 01:14:57.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"c2b9cb1014ecea1be4847004b08daf33fb23ca43","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-27 01:15:35.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d8a147aca20a7679352f7b8944c5e610b528802c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-28 01:58:05.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6f44b18b2ea68d154b13af0e0e6ca34e911b955b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-28 03:09:30.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"056f11e9a70e3c52c7e15216c1b561a46890ff39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 03:24:24.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e1a517ce0b463e92fa020a921fb4ce\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81513ffd4fbd488fbb21369fa0e97bb1 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29a43a3bafd54fc4a1c930a665e9780a : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6135579bb244a26aaf041b348a30947 : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6126ceddfa67122b28a3cc80bce5cc55928c6ec8","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-28 03:32:28.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"571639ccd5e49bb04a4d2ba126f448e56519567e","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-28 03:33:01.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"f5029cd18489ebb1fea941d0a81da8e8ec79b3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 04:50:30.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b90519485714776adf575177f135b9e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84e5ef7c431e4f3d856396034f24c468 : SUCCESS in 5m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cd96b9f0757483d9feb0e57c9b0b235 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dbc53e4aad649c99e1c7dc9981ff896 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25060b25bc294025ba4cb0e6f070687d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30f6aba15d764cebba670d169d235ace : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b31403b3ba04db88ee6e212b2d95b83 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7a22d03d8b42aeae6e43a8af6aaea8 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42bd3e185f1b42a191af5ca9a0d126d0 : SUCCESS in 6m 30s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/80b093efde3b4f5cafa5e21da66a127e : SUCCESS in 1h 39m 03s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ea40ee6437ba4d41b6f36213f6e8fb52 : SUCCESS in 1h 29m 04s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1057a30229014c4b8ec90cca52088636 : SUCCESS in 1h 34m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/479bc4bfbfd14677b743c37c38083dad : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":18},{"id":"28cc85a91debd3c10a9b68c06da0ce76ef65d9f0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-28 06:34:43.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c459497d32d53f52f4bd837b10a9d850ffec6f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 06:35:03.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"1748965421dbdc4e61e9577d0aaac18a6419757c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 07:09:51.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7d1fa4b7fd44af92526526a9a26291\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e636aa5f47384e079bc43395b52c4504 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4b1ad52db1043b5a8ca6d19d327308e : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec821fbe4f024a74897ad5f6c917cf65 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/841c821d0bfe4072b19475ad63b15864 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c468e719ba0543878f74b52005b83885 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1926162f56a8462dbc886912f39c5e7c : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b04f3061a6074f9a8057e20768f78f8b : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":18},{"id":"c8815c45b6b0fa54b72e6b8efb5817f616ee7562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 07:09:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"0a1be2418d646b234df73049995ca28acc9f8f61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 07:10:53.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/729863c20f4d4999b4f6c25d8e691c87\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c983bb6110b5443f95465acd960cdc35 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c1188328164ef0badd86763438502b : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa52f13939144dc5a635d8f4c7a863ff : SUCCESS in 41s","accounts_in_message":[],"_revision_number":18},{"id":"31e0e139bbc98701d6c408be838128afce77ac3b","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-12-28 07:13:29.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"40dd8e9bbd6fee1f96242519159f3e938f6eab9b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-28 07:57:02.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"1126317f87f0c2dac74a8b6c263d38eef4130ab9","revisions":{"4bf75ba60bcedb68df6bc54432bc3d81631babd9":{"kind":"REWORK","_number":1,"created":"2022-10-18 17:14:42.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/1"}}},"commit":{"parents":[{"commit":"346e1612200563cdbeadc49186fd819dc6d4f950","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/346e1612200563cdbeadc49186fd819dc6d4f950"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 17:13:08.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4bf75ba60bcedb68df6bc54432bc3d81631babd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4bf75ba60bcedb68df6bc54432bc3d81631babd9"}]},"branch":"refs/heads/master"},"12ba644ea3b2283c647b5a0f1ea28ac41a1e0084":{"kind":"REWORK","_number":2,"created":"2022-10-23 08:35:11.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/2"}}},"commit":{"parents":[{"commit":"346e1612200563cdbeadc49186fd819dc6d4f950","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/346e1612200563cdbeadc49186fd819dc6d4f950"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-23 08:34:39.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba644ea3b2283c647b5a0f1ea28ac41a1e0084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba644ea3b2283c647b5a0f1ea28ac41a1e0084"}]},"branch":"refs/heads/master"},"7ca27e4ce8d8452dc8d81a5a070a725784b56175":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-23 08:48:09.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/3"}}},"commit":{"parents":[{"commit":"346e1612200563cdbeadc49186fd819dc6d4f950","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/346e1612200563cdbeadc49186fd819dc6d4f950"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-23 08:48:09.000000000","tz":0},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nStory: 2010004, 2010346\nTask: 45128, 46489\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ca27e4ce8d8452dc8d81a5a070a725784b56175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ca27e4ce8d8452dc8d81a5a070a725784b56175"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d41c957410d20d2a8ff6a7e0f30c64dba8a0dae2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-10-23 08:49:04.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/4"}}},"commit":{"parents":[{"commit":"346e1612200563cdbeadc49186fd819dc6d4f950","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/346e1612200563cdbeadc49186fd819dc6d4f950"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-23 08:49:04.000000000","tz":0},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nStory: 2010004\nStory: 2010346\nTask: 45128\nTask: 46489\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d41c957410d20d2a8ff6a7e0f30c64dba8a0dae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d41c957410d20d2a8ff6a7e0f30c64dba8a0dae2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4b234af0a1fc0c3d7f342c3fd63bd58e5b6634ff":{"kind":"REWORK","_number":5,"created":"2023-11-27 16:30:16.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/5"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-11-27 16:30:03.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4b234af0a1fc0c3d7f342c3fd63bd58e5b6634ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4b234af0a1fc0c3d7f342c3fd63bd58e5b6634ff"}]},"branch":"refs/heads/master"},"4a63e3aebf5be4b419617779b4f1228ea5b3099e":{"kind":"REWORK","_number":6,"created":"2023-12-07 09:43:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/6"}}},"commit":{"parents":[{"commit":"aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9","subject":"Merge \"Remove unnecessary ceilometer service overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-07 09:42:09.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wuchunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a63e3aebf5be4b419617779b4f1228ea5b3099e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a63e3aebf5be4b419617779b4f1228ea5b3099e"}]},"branch":"refs/heads/master"},"26b58d6f4467e131ff222ac5ac330bafacc6a4df":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-12-07 09:46:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/7"}}},"commit":{"parents":[{"commit":"aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9","subject":"Merge \"Remove unnecessary ceilometer service overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-07 09:46:20.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26b58d6f4467e131ff222ac5ac330bafacc6a4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26b58d6f4467e131ff222ac5ac330bafacc6a4df"}]},"branch":"refs/heads/master"},"159d70ac00b7000108363056b221c865e384c01a":{"kind":"REWORK","_number":8,"created":"2023-12-08 09:07:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/8"}}},"commit":{"parents":[{"commit":"aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9","subject":"Merge \"Remove unnecessary ceilometer service overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-08 09:07:10.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/159d70ac00b7000108363056b221c865e384c01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/159d70ac00b7000108363056b221c865e384c01a"}]},"branch":"refs/heads/master"},"9173faac6dd81e1f2c6ec94c8f11c72dd48ba380":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-12-12 06:58:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/9"}}},"commit":{"parents":[{"commit":"9f7b9a95e86d69fce9ada96c16677d1e1b168884","subject":"Revert \"Fix docker daemon failed to set IPv6 gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f7b9a95e86d69fce9ada96c16677d1e1b168884"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-12 06:58:03.000000000","tz":0},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9173faac6dd81e1f2c6ec94c8f11c72dd48ba380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9173faac6dd81e1f2c6ec94c8f11c72dd48ba380"}]},"branch":"refs/heads/master","description":"Rebase"},"ee3236ec806a2512e22da2d2bc80aa821c81edcd":{"kind":"REWORK","_number":10,"created":"2023-12-14 06:54:07.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/10"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-14 06:53:51.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3236ec806a2512e22da2d2bc80aa821c81edcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3236ec806a2512e22da2d2bc80aa821c81edcd"}]},"branch":"refs/heads/master"},"2e93cb6e9d21f753b48d0a2962931505cf954441":{"kind":"REWORK","_number":11,"created":"2023-12-14 07:16:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/11"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-14 07:16:35.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e93cb6e9d21f753b48d0a2962931505cf954441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e93cb6e9d21f753b48d0a2962931505cf954441"}]},"branch":"refs/heads/master"},"ef5ca9c50668aefbfe4fb0643c33156c3452611b":{"kind":"REWORK","_number":12,"created":"2023-12-17 15:11:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/12"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-17 15:11:02.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef5ca9c50668aefbfe4fb0643c33156c3452611b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef5ca9c50668aefbfe4fb0643c33156c3452611b"}]},"branch":"refs/heads/master"},"7c89017401a65dcf6b035fcac80d222c4de1b186":{"kind":"REWORK","_number":13,"created":"2023-12-18 01:25:58.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/13"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-18 01:25:50.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c89017401a65dcf6b035fcac80d222c4de1b186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c89017401a65dcf6b035fcac80d222c4de1b186"}]},"branch":"refs/heads/master"},"5946e55dec0fb4d3991c1c8134685cb3dd2a630b":{"kind":"REWORK","_number":14,"created":"2023-12-18 09:26:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/861781/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/14"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-18 09:26:41.000000000","tz":480},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5946e55dec0fb4d3991c1c8134685cb3dd2a630b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5946e55dec0fb4d3991c1c8134685cb3dd2a630b"}]},"branch":"refs/heads/master"},"bde508bc34968c0c0d6381dbb4a8ba2a56e01c48":{"kind":"REWORK","_number":15,"created":"2023-12-25 02:03:52.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/15"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-12-25 02:03:27.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bde508bc34968c0c0d6381dbb4a8ba2a56e01c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bde508bc34968c0c0d6381dbb4a8ba2a56e01c48"}]},"branch":"refs/heads/master"},"a44bbdb5d0260ffc55603992fef86b0db3689118":{"kind":"REWORK","_number":16,"created":"2023-12-26 02:17:43.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/16"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-12-26 02:17:20.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a44bbdb5d0260ffc55603992fef86b0db3689118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a44bbdb5d0260ffc55603992fef86b0db3689118"}]},"branch":"refs/heads/master"},"ea87734ca29b5f7628826fffda53677d2319bffd":{"kind":"REWORK","_number":17,"created":"2023-12-26 09:43:10.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/17"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-12-26 09:41:33.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ea87734ca29b5f7628826fffda53677d2319bffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ea87734ca29b5f7628826fffda53677d2319bffd"}]},"branch":"refs/heads/master"},"1126317f87f0c2dac74a8b6c263d38eef4130ab9":{"kind":"REWORK","_number":18,"created":"2023-12-28 03:09:30.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/81/861781/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/861781/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/861781/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/861781/18"}}},"commit":{"parents":[{"commit":"895e59237efb17f8d46763e58210dae96f772d7c","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-18 16:54:24.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-12-28 03:09:10.000000000","tz":540},"subject":"Disables shell\u003dTrue in backup and restore","message":"Disables shell\u003dTrue in backup and restore\n\nThis PR disables shell\u003dTrue in backup and restore.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010004\nTask: 45128\nChange-Id: I13405e8bc93586441da28898e99dce936fc6d68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1126317f87f0c2dac74a8b6c263d38eef4130ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1126317f87f0c2dac74a8b6c263d38eef4130ab9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
