)]}'
{"id":"openstack%2Ftrove~862784","triplet_id":"openstack%2Ftrove~master~I41ebe8cf844c24f9eb178f232bd66af5ed7298e8","project":"openstack/trove","branch":"master","topic":"story/2010242","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-01-27 14:54:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I41ebe8cf844c24f9eb178f232bd66af5ed7298e8","subject":"Fixes the error when stopping a container","status":"ABANDONED","created":"2022-10-27 08:15:42.000000000","updated":"2023-01-27 14:54:56.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6608ea1085ec46ec82c9178e68200541d8184900","_number":862784,"virtual_id_number":862784,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-27 09:07:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-11-01 14:33:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 08:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-27 09:07:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-01 14:33:48.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"6426021b879173a1a7001b04380c9050eddcf32e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-10-27 08:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"719c24ac3ccf692b40f4d3b6d10932b58a938861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 08:37:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02320cdb5ab4b5a9bc4d0a0ec67b72f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a46eb88582864939afa115a51f4b0595 : SUCCESS in 16m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2fe098cdb4146348d540290aebbf836 : SUCCESS in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b3f984e5352881de1c2c4292e9e6662aa3a8036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 09:07:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa6ef442554149c7b44e8a703718a0e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/506e3435d9944528bc6c2d6786cd0a4e : FAILURE in 6m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e6e8987ccf40b5a197e9e56337fe68 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1758a9bd1f1446c08ed030eddcef2de2 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7bb20497b4d47e38d3f16a4c758b0b9 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a970270c0464ab79e49b0ba4561af9a : SUCCESS in 5m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d904501473354a299311b0a5e2552514 : SUCCESS in 5m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1a4dc49b8c7048309b053a3a45fb4bf3 : FAILURE in 39m 44s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/923dcc16bc714f3fbdba6413a48629f4 : FAILURE in 29m 46s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d713a05fce474d92824351a5caf07f18 : FAILURE in 40m 06s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/004ff7a2906540fc9d57a9aa4fb23c53 : FAILURE in 26m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd21453415cb2ff1759c40da9425a53727f73f55","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-01 14:33:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6608ea1085ec46ec82c9178e68200541d8184900","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-27 14:54:56.000000000","message":"Abandoned\n\nI will abandon this because it\u0027s not a critical one.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70c182bc43904ed7ac2b74744506a2a67d59581d","revisions":{"70c182bc43904ed7ac2b74744506a2a67d59581d":{"kind":"REWORK","_number":1,"created":"2022-10-27 08:15:42.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/84/862784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/862784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/862784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/862784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/862784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/862784/1"}}},"commit":{"parents":[{"commit":"3789a3951b120ebd0eaaa5a9b196d3cc30418092","subject":"Merge \"add openstack-python3-zed-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3789a3951b120ebd0eaaa5a9b196d3cc30418092"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-27 08:11:49.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2022-10-27 08:14:40.000000000","tz":540},"subject":"Fixes the error when stopping a container","message":"Fixes the error when stopping a container\n\nThis PR fixes the following error when stopping a docker container.:\n```\nUnboundLocalError: local variable \u0027container\u0027 referenced before assignment\n```\n\nStory: 2010242\nTask: 46055\nChange-Id: I41ebe8cf844c24f9eb178f232bd66af5ed7298e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70c182bc43904ed7ac2b74744506a2a67d59581d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70c182bc43904ed7ac2b74744506a2a67d59581d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
