)]}'
{"id":"openstack%2Ftrove~865221","triplet_id":"openstack%2Ftrove~master~I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e","project":"openstack/trove","branch":"master","topic":"trove-local-registry","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-02-27 16:11:02.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-02-26 13:17:36.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e","subject":"Using local registry in devstack.","status":"MERGED","created":"2022-11-22 03:19:42.000000000","updated":"2023-02-28 07:43:35.000000000","submitted":"2023-02-27 16:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865221-trove-local-registry","meta_rev_id":"6a448fe3c95c423adbc208fc7e2669bc065bab87","_number":865221,"virtual_id_number":865221,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-27 16:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-27 14:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-27 14:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-22 03:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-22 03:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-08 03:34:16.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"7460b9b0f1047ab745e693843f11675331bf9085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-22 03:19:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a87fcf86da25fb98b79ebdf1e576a5c910c6676","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 03:35:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a663752bf37f4f77bc1838715b5b7950\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b53e27690a8c40fc84b437f56c0889c6 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6bf6ab1a81874c3dbe6704acac9a6e7e : SUCCESS in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fcfbcfdc54af5bad13443dc87820b40c7154df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 03:48:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6416995f291c4c608d6db7a0e2e52445\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97f022c899414c26a8745f801a86b83f : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d1085c386647f680fb995e49d5b70b : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ee3283449a948b4b12665946472cfef : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06a944ee79fc420f8b603fd07cb4c596 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6918a5a443748dabb6783ad676832c4 : SUCCESS in 8m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4945855cdaab42a4aa7f0c5d5f308676 : FAILURE in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"548e582af4ba20c388c925da128b0f1745b1398d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-03 14:13:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba6f1e0ec77c6cd1f68984d4fca6d66c10d094bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-03 14:32:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc9bd940e285478f913839e9d0563772\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e45f91675bdd46c4bc8cce6cd73a4dd7 : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e3612b021ab40ad8d6295aa5f70da11 : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68b3c21e6fe8b058f3939347fd1a2807cf23a918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-03 14:58:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97eb13ff9e4f421f9feb0547df5ee6f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac7e84517f2d4ca0bdad1339bb2e8787 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3a28995ae0418f9e95684342b601b2 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01fa4db2d1324ab4bc4f17e7bb75c2e1 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c8d066253b4abe80201d063cab42fc : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2370552f4942b595ade115f2bb57c4 : SUCCESS in 7m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/46ec9a69f16749c0a745d1e3dd1abecb : FAILURE in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d673c2fe1438e0fd23393eb80230a81725257315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-04 13:33:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db28bebf5fb79352ab8ad6948a6db1727ad28722","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-04 13:50:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38d73de96331476c8cf38645e4976766\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb870956fbac46cf8188219d0e8932a6 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca3ea0ba8b0b4ccbbb94e47cb96f2fb4 : SUCCESS in 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"748f5de22eb8bb4fb9314a16e5754eb658e15b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-04 14:08:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94dd52d7818f4a8a954d7c94b985bbf6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0456a2223946412d852c98fb546313c6 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905ff0d1e898432999ce6cf94691317f : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5159203664df4fadaeb3d142e3c28733 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ea705f83d6745e3b7b1a68935957ee4 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2692ad5e45044e0b688f9d076ef0ebe : SUCCESS in 5m 02s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/89324daf66af4502adb5b90555db9838 : FAILURE in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3fd4863e349f9f4a0840a9d37ed8728211099c49","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-08 03:34:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7836be11815f0d6a72b88889e03e179566d79489","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-12 08:39:09.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"c62f71ba72b5578c7cadcb12e4be0120c1391ff4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-12 11:37:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6d2546c7e39bf3424cfaf6cce90bb4c3319a975e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 11:52:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a31f41e224f448983bbdd17b3993c5f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4eaeca6c6204f47bbe50760fd41abeb : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c442cfdf00b4876a141dfacd570cefb : SUCCESS in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"86fd465acf2652d8a05328fa9fb1667e3df5219b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 12:14:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e126e04fec5043378e3e9933d1114c02\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60729bdd39fe41b8a915bba34f7c4b0c : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22bece13cc9e43d18f9d51a2c7b992bb : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f73728fe741429d842f0ed2e405e6b2 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3170c93733d249dbbc35d8dfa4ab69d0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2ee155484042f491ac52fc7ecf624f : SUCCESS in 5m 37s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/970ef29770dd48479092f02ecc3a1a13 : FAILURE in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0be68936f2256316f604a81d6d7c3e626ae6153","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2022-12-16 03:36:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee25498de7b773d4dab0f97d409e9ca98298b5ba","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-29 07:05:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b0e4ec84e92a65727332d4bdbd0d13aa140503e7","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-29 12:32:04.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"752160faa597943644c26a800b44a5773d447f05","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-30 03:49:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"39cab84dd51abc7c9bc610ab8b9bd15ce1b546d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-30 06:43:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"898ff2f9e959090426a16d233f91c8c885bd52ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 06:51:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6606db84af4d309fbfc277f8058a02\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da87cbcb8d9d4f8c823e21a1da88c9c1 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3af1997af9bb41c4b8bfa869d523784d : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cbd901fd5e0a81957319ae47aafea724c4071e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 08:30:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a0104862d7b4b969c17365e0b927a5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04cc45dee3d848cbb78307c97761368d : FAILURE in 5m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424dbb8f61a24cb1a6b3a056c933292c : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b141fcb9193e4a51b264edb9d70d1224 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d4c461efe9a4a6d81050f975212665e : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153f05d4f2534530b9c0bc9600b526c5 : SUCCESS in 4m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f878edac1394f0bbd2d7fe07b3b0340 : SUCCESS in 5m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9a042574a68f4c08ad5247913bb0eb3f : FAILURE in 51m 15s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/cfbc35b561454e318ec9b6c80fd93d05 : SUCCESS in 11m 40s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/5ec74e059ede41d898988af3847f4cdd : SUCCESS in 13m 19s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1056d788810a45b38706769df01c2e88 : FAILURE in 49m 12s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bed464915254b5f931af3a7a0873dc6 : FAILURE in 1h 21m 46s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/df0ffe7d43bf4ba482960857c4ef2a47 : FAILURE in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a0d7bf8436f3fef924beca73c5d36fc9034dc410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-07 15:55:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"85892eedf0b2ae76b4fc77f7cb65e40419c55cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 16:02:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2030239b1242427d83c761609be8e23a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b6de0f8748447cba1745719763ecc80 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da9c08d7532347bcaefc22ca9aca9c4e : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"17e985cec752393c8e88d592bcc0d64c0bc08ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 16:57:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b08b6531d40c4e848b0ce8133eac9360\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60666a6d52664fcc9d65aa2d6fbdf666 : FAILURE in 7m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ddefee3f8240969de37e16df3cbcd6 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d1d14cb2ecc42beb2010b7485d11c33 : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d22ea7e37ee743258815c98a321a2b83 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887f1f4ea5064dcba88874ac146adddd : SUCCESS in 5m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8c900ac80234b26926ddc1cdd7897c3 : SUCCESS in 6m 25s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/45d785e229bf426c8eec5c0db946afb1 : FAILURE in 57m 43s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9e1de8cd71594452a3d8a91562859b6c : SUCCESS in 8m 04s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/c1c86d450da049ac9894e869a8f7cbbb : SUCCESS in 11m 01s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/97c9d91d969b4057867c88c965737615 : FAILURE in 50m 52s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cf31ce1f7dc453fa47b7085e0588ad8 : FAILURE in 59m 22s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/815fd1cbb77c4543bb9024783c39af81 : FAILURE in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ab801ab7f51f5b4060eecb4d7a893de2b52aab3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-08 01:18:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"042dc6513ead864a6f0a2a133520f86f35679b2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:34:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1b511a28ad74ec4a569ec0d3f839938\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31a08aef12634c2f80bfa4eb177f114e : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06d25bd17554424ebfb0b4eade08c7dd : SUCCESS in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0b5aeebc4393a6375d94bb94f089ab750f4c7646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 03:18:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cf1c2a86a244c5e9cdc47cb66bd2f80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44005085be664ec982acf668a9eb9e2f : FAILURE in 5m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e05f52f6b44d5e87a1ceaa8535cd0c : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df8efdbc8ba46fd8bfb92b5b3f3287b : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ee3cd2e64a644deaa3b5e83a634fa70 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4aff73e4ac4473fac267a498efe7778 : SUCCESS in 4m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/074f53ff26ad448782717506ba0bd6ae : SUCCESS in 4m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/26e4c94e58854c199a292857167efcf1 : FAILURE in 49m 55s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/51f1c11a23df4d6a8fa8a26795a7d44b : FAILURE in 5m 11s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/ca008cd1525c4466ad6fb2c2f0ceb9ce : SUCCESS in 11m 34s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/951e441b4ddc46058ed4072d2326622c : FAILURE in 49m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/301d98e220c6445ca0b22d6410ff274c : FAILURE in 1h 43m 22s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ca9f1e28775547249aee7c0743d645fd : FAILURE in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79b6f74105bdfb8d30324e6825b8bb850e631f2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-08 16:18:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e1d4854484ccce60a3e691f06605cc26ee082a0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:26:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02593b9a23b846009aa47e8f53f59a55\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f51020022e14ca5852559cc132784cd : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e133cc683354a1b9f25b7394f53c3c4 : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d5513fc9b32b878b3b180eef88081dbfcf572798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 17:32:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21cc32c47db142968ee0f5796689e1ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce4f069674c43ea9e353a2cbca2d22f : FAILURE in 7m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b6ada7e93ef4c349b39eb88cb1d26e7 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe7fadaec62c4252ba4416e17121ab1a : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4264a2ad0a474987874594558c8199c0 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70239b6cc96c4a439e4ee4bb6ac48da4 : SUCCESS in 5m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad0ba779083541fd9830fef4387c7b81 : SUCCESS in 6m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/acb13caa20624a64ae843ff3469b8a3e : FAILURE in 50m 53s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d10565f65f9a43529e4165d3a34c49cf : FAILURE in 48m 53s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48710d648e934baa8fe5b88e779f3b11 : FAILURE in 1h 10m 25s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e2f24e26312b46fc835baae897578e4c : FAILURE in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1e1f892bfcc93ff52d79617405b0b2ef54670e07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-09 15:21:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dbf3507351bee3dc7454e003e96a12777d9118a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:28:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bab6685612648ccbcf12420dd975058\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9398e3cf24044a9fae36075bdd0e0761 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea6c844beebf411c86a85456603da7cb : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9835509ea25437412cc2e02eb5bb3d3f9fff12cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:39:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f088ba4e6da94751b99e333737cb8af8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aed1c6849d964f079cd238701718e95e : FAILURE in 5m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8904a4da275346d78a7e3340eef8b9b5 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72d82c06d5bd4d989312ceae9809ef83 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2938ae0ec8564d789271c548940b2f69 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb6af5708754748a1d6d3ef10707aa8 : SUCCESS in 5m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fed562eb792947f29d484e0dca3d47ec : SUCCESS in 5m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cdf44b21a6fe4ae2b8664bdc6c3e4e68 : FAILURE in 51m 17s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a93d3408b6874358ab81e025c42df494 : FAILURE in 51m 06s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1166cb9ce8cf483091b84f94e4e22f8f : FAILURE in 1h 13m 10s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8c6976a0d16d4f1d8e1e7ac8ff710645 : FAILURE in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dcf991ece1190cb8fced9ff257d71db424d8bdca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-17 13:35:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fafe1792b56282de4ba39237504618daf69ff0d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:43:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a617af4bf048de8255dff54e289491\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd89eea2cda842139e7f98c27f528a45 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1cc151b853be4b468323786455f72a6a : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5666ae65614575d28468f7a7c3f953ff573c3e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:56:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f9558fccaba48b2aaede9351b9104de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87b2db07c3ab4edab68677e6ea2b4139 : FAILURE in 6m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7fdddeea6514e6cb584e4823e335bc5 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51b4865b209a42faaaf4d4f91251bdce : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48b9194116cd44e9b6385e7fa35cd38a : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a70fbbf523b45bcac8351e2602eef51 : SUCCESS in 5m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46b61c9db58141e0a5ba9a8be384ac16 : SUCCESS in 5m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/037623302a3d4c608615bab6a5a0eabf : FAILURE in 1h 07m 34s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7b007aea4f67460682bdb1072ebf584f : FAILURE in 1h 09m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4ebe0c6d466454da854e0d5e3479d9f : FAILURE in 1h 18m 20s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/92780ce3d7054bb2ad21a6f5b9829545 : FAILURE in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0eb9afd3d980c85edec35def4b48a3778cb72498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-20 14:05:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"35d1699f79071503c34bcd69aefa9cd944e4cff6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 14:59:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ac9223c7bb416a9691f931f82bfdb2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fcfcc0fa8244533bc4313b2534a4afb : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/51896966f1ad452e85def699c21a264d : SUCCESS in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"31a310247c26d127bec80c207e919c7a5942d91f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:23:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75acd28e7484426f96799e89ab726b81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/495e2e3065c044c986c2650638f6ffab : FAILURE in 7m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee878288305b48999c38b68dc35adda7 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5eee611d6a749d99f59dbff0fb6d276 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d087baeba1da4e4c97b527e9600150b7 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b95309089424323b26dca75c866b9b6 : SUCCESS in 4m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/185feb4737e84789afd3e4b89baff734 : SUCCESS in 6m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/877f49797ecf459790796a2a57384bfc : FAILURE in 49m 48s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a751b8e55ac746c199baf499cc691b54 : FAILURE in 51m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bfb5c2eb5ab497682114abceecd436d : FAILURE in 1h 14m 40s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/46a3070b33294a78ba0c58e70222fd12 : FAILURE in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0fd465047a552047fcb6523951a88ee9bb668056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-21 04:09:33.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"21fdd6a94e3afd575cf8d70dd752603adc4f7232","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 04:23:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14e6c73d43b64e5f959b84774d24992d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b3964cabbac4845a896fb5171e55bce : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8bedb7a33d444aee99c787f4572b0d2a : SUCCESS in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e4a94c014a19bc72f952f3adb4be9a37dfd141ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 06:13:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49aa2f522fe7424a9926ef4b4380f231\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0beb2eda2d1240229a56d3b297aaa443 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a809d9c64ba64f2ead708496661458ba : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/091b228d140343a1a5094e0f4dba960c : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e687251295bb4a1fa2a2bcdefa485620 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadd103c2ffd4c5a84e39db3e1f25e04 : SUCCESS in 5m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37d8d021dc2744cc9e91107191f647a5 : SUCCESS in 4m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f0872be3fe694cf481927b5938596c96 : FAILURE in 1h 48m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/44c1d942411e4bbc968df34ea7359534 : FAILURE in 1h 08m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/667d9fd77c2949acb8b8f86d8fd58769 : TIMED_OUT in 2h 02m 29s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/39a9633022f34931baa6c5b1bf48176b : FAILURE in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c987bae27a5ceba0195a3be15635dc1bbac8d922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-26 09:27:35.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8f57d7a7e17f25d5331a7d20f4392eaf166e2bbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 09:41:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/764fe2bccf5a4a65807c1e5e029323c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67cc7ad536814852a63bfd969f962489 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/899f920ae20647c386dd8dcabc58b63b : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0096535fdee571d6fb4daa21f54ab3a964d4c610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 10:00:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd94875914e3452792ab495d768fc587\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b142cd306df14f94b0862c62e54d2cfe : FAILURE in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f15e59ef3443a981dd474f818ad1d2 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/709041aada7f40d2ac0805629fa78868 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bbaa6a0e3334f02baff670a32b02c67 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c45f2652fc1470784a1ea024410a412 : SUCCESS in 6m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8686797cf1b5433a9ea74d146c5d3e74 : SUCCESS in 5m 08s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/71ddf6fb431c475abed7d58a71f28a8d : FAILURE in 27m 03s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d2cb9f84828244fd89a298c2bfb9b900 : FAILURE in 28m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59368854f8154dacba70ecebab930212 : FAILURE in 31m 20s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0c0d6a48f81b489ea3e9bbcc9b2915b3 : FAILURE in 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"31ffb4f02ff1a56646cfc7e088f4c5e3a467ca0e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-26 13:17:36.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"44fde5cbc3c14310c6c2cd47f09d4bc2220a9000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 13:35:07.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d62ebd6c401046f2b9815d787b1182bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98d95d0d992a432fa7ab51cc4565eb9a : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/676db504c5fc4d429b615e4dd6d077d0 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9782fc9b545a0e05a6e27736fb2bab28066b3547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 15:23:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a66620d441e34471891d07f19d7ce233\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/593f97caa77245b882cf1cd22ddd6ba5 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3ef2146bb054f4d91370114e6b4570f : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/333b15b8ac8f4cbf8ef0adb16eab415c : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/765642f7315447649a02b04e739dce84 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467619c29b9241649ade204af97042ca : SUCCESS in 5m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71b67eaa1c954464b07b548176624471 : SUCCESS in 5m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3fb593635b67427c8302f8465b09b340 : FAILURE in 1h 59m 00s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/fd02798782c34238966a8395e2c9118a : SUCCESS in 1h 51m 26s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca811800aeee4b5da499233e2b3c8703 : FAILURE in 2h 01m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/583b232e124c459fb28343d903d19704 : FAILURE in 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a7e225eef6879aa651bea60c653a756bfb049ae7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-27 14:00:14.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"38e0f7d6139205f4b76400616d06216181fb0e36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:00:49.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"2a75e451597338c0e0d3c1d89d4d3cc786079742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:11:01.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a0e8083b9348e589ae9d06e7721ece\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f9ab9bb4e643179197df05a8aec1f8 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec0551a8c5a341cebf2fa9ed8838a0d5 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc5c6e71443942a985a55d15f2e00f08 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878f45b04eec4c8b8e42be96cb6a8e3b : SUCCESS in 4m 51s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cb099500edf34511b3dcda9ad539c857 : FAILURE in 2h 08m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9d467bdf462ec7d814d2f56e2f5974acb106de79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:11:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"a1676d794bdba80293278aae57863c8fc79db97e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:13:02.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/603637c96d5c49708a96482379f5a764\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d120ce8ea1d6405ca81a9785c8be2c43 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":13},{"id":"6a448fe3c95c423adbc208fc7e2669bc065bab87","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-02-28 07:43:35.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"1f1e2becfeecb3a7e061898890c49821eef95672","revisions":{"af1707f61b0498966058d99036a30bc88448a849":{"kind":"REWORK","_number":1,"created":"2022-11-22 03:19:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/1"}}},"commit":{"parents":[{"commit":"5a960b37a463984aea3fff078e8fcb5ccd85a949","subject":"CI: test build docker image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a960b37a463984aea3fff078e8fcb5ccd85a949"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af1707f61b0498966058d99036a30bc88448a849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af1707f61b0498966058d99036a30bc88448a849"}]},"branch":"refs/heads/master"},"cdc55ef816195fb3e4d338366fd6bcefdb705dbe":{"kind":"REWORK","_number":2,"created":"2022-12-03 14:13:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/2"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-03 14:12:41.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdc55ef816195fb3e4d338366fd6bcefdb705dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdc55ef816195fb3e4d338366fd6bcefdb705dbe"}]},"branch":"refs/heads/master"},"5dee17c007a522c8e0eb8a25584a4dea101b1ebe":{"kind":"REWORK","_number":3,"created":"2022-12-04 13:33:58.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/3"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-04 13:33:32.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5dee17c007a522c8e0eb8a25584a4dea101b1ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5dee17c007a522c8e0eb8a25584a4dea101b1ebe"}]},"branch":"refs/heads/master"},"3ea25e740922ae76ea21edd629f10dd266699ccb":{"kind":"REWORK","_number":4,"created":"2022-12-12 11:37:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/4"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-12 11:37:43.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ea25e740922ae76ea21edd629f10dd266699ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ea25e740922ae76ea21edd629f10dd266699ccb"}]},"branch":"refs/heads/master"},"99e38879f388b2b29b10f37da37c08213958e8de":{"kind":"REWORK","_number":5,"created":"2023-01-30 06:43:31.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/5"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-01-30 06:43:09.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99e38879f388b2b29b10f37da37c08213958e8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99e38879f388b2b29b10f37da37c08213958e8de"}]},"branch":"refs/heads/master"},"a2139e7120eadaea77b7072d9e8c30a484609fd2":{"kind":"REWORK","_number":6,"created":"2023-02-07 15:55:23.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/6"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-07 15:53:26.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2139e7120eadaea77b7072d9e8c30a484609fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2139e7120eadaea77b7072d9e8c30a484609fd2"}]},"branch":"refs/heads/master"},"5280b59289a14bba13706883889688e30abc0c51":{"kind":"REWORK","_number":7,"created":"2023-02-08 01:18:34.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/7"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-08 01:18:18.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5280b59289a14bba13706883889688e30abc0c51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5280b59289a14bba13706883889688e30abc0c51"}]},"branch":"refs/heads/master"},"347c1440bc92d34e47eeb301386cbce9e0317828":{"kind":"REWORK","_number":8,"created":"2023-02-08 16:18:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/8"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-08 16:17:50.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/347c1440bc92d34e47eeb301386cbce9e0317828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/347c1440bc92d34e47eeb301386cbce9e0317828"}]},"branch":"refs/heads/master"},"c462fa226e867bc911333f35f1f9909b1a6aa89b":{"kind":"REWORK","_number":9,"created":"2023-02-09 15:21:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/9"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-09 15:19:15.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c462fa226e867bc911333f35f1f9909b1a6aa89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c462fa226e867bc911333f35f1f9909b1a6aa89b"}]},"branch":"refs/heads/master"},"c93e844cb41b27fd6ef817d36704c942032cfd79":{"kind":"REWORK","_number":10,"created":"2023-02-17 13:35:45.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/10"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-17 13:34:57.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c93e844cb41b27fd6ef817d36704c942032cfd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c93e844cb41b27fd6ef817d36704c942032cfd79"}]},"branch":"refs/heads/master"},"69169599bc69123d46d8097177a969f3053d0ae3":{"kind":"REWORK","_number":11,"created":"2023-02-20 14:05:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/11"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-20 14:04:22.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69169599bc69123d46d8097177a969f3053d0ae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69169599bc69123d46d8097177a969f3053d0ae3"}]},"branch":"refs/heads/master"},"adf3c6ed960a5a936a9aa57f494e80535ec79c24":{"kind":"REWORK","_number":12,"created":"2023-02-21 04:09:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/12"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-21 04:09:19.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/adf3c6ed960a5a936a9aa57f494e80535ec79c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/adf3c6ed960a5a936a9aa57f494e80535ec79c24"}]},"branch":"refs/heads/master"},"1f1e2becfeecb3a7e061898890c49821eef95672":{"kind":"REWORK","_number":13,"created":"2023-02-26 09:27:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/865221/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/21/865221/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/21/865221/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/21/865221/13"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-21 14:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-26 09:27:17.000000000","tz":480},"subject":"Using local registry in devstack.","message":"Using local registry in devstack.\n\nThis patch runs a registry container during the installation of\ndevstack\n\nChange-Id: I5e5a89cd9c65d1ea5b47fece985dbc044bfdec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1e2becfeecb3a7e061898890c49821eef95672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1e2becfeecb3a7e061898890c49821eef95672"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
