)]}'
{"id":"openstack%2Ftrove~868917","triplet_id":"openstack%2Ftrove~master~Ic1c2d0e100faee0fe57fde01a49190cd69c6a532","project":"openstack/trove","branch":"master","topic":"CI-build-centos-image","attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-12-30 05:41:46.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic1c2d0e100faee0fe57fde01a49190cd69c6a532","subject":"CI: build centos guest image","status":"NEW","created":"2022-12-30 03:32:08.000000000","updated":"2022-12-30 05:41:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b8f2f940661b0be3f4f1dbb8dbceef912534e91","_number":868917,"virtual_id_number":868917,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-30 05:41:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-30 04:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-30 05:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5813dfc71f9c4ec8588dcdeee240458ea37d1f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-30 03:32:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"813adfedd481799a8eb3efc863247a869f7a6d5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 04:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/329468f3b42548859113bffbf0ca476b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07d3748cd3ec47bcb211ce7e3fb85bda : FAILURE in 13m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d1e777bb03841fd8fd577d389657bde : FAILURE in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b8f2f940661b0be3f4f1dbb8dbceef912534e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 05:41:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d41120673304f7f8f95f8f2fa6bffca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ff357f91d4c44618ad8ed150586d968 : FAILURE in 6m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b065f8f62b4f0988fdc1c415434347 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c83daf7e96be480b9ab24d0739e1ccbb : FAILURE in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08fe31624dc84ebdb3746a83fb961b7b : FAILURE in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651f6711cb5f46cd97b436f3f63be276 : SUCCESS in 6m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a6eb2e4cdf642b4a4c44902f94bf874 : SUCCESS in 4m 37s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/47686ea78b3749519dc1e35544476780 : SUCCESS in 1h 35m 18s (non-voting)\n- trove-guest-image-centos-build https://zuul.opendev.org/t/openstack/build/150d352970ec4159b0f68c0815877836 : FAILURE in 2m 59s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/f2f01045cd6c49029dcf9737e1fc158c : SUCCESS in 1h 36m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1366654699bc4934a01fb60395bb6326 : TIMED_OUT in 2h 02m 31s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6ece9d0232d544619870c5322c08260a : FAILURE in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"618edd5e9657dcba2959f42083bbd15836001f87","revisions":{"618edd5e9657dcba2959f42083bbd15836001f87":{"kind":"REWORK","_number":1,"created":"2022-12-30 03:32:08.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/17/868917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/868917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/868917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/868917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/868917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/868917/1"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-30 03:30:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-30 03:30:22.000000000","tz":480},"subject":"CI: build centos guest image","message":"CI: build centos guest image\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/840113\nChange-Id: Ic1c2d0e100faee0fe57fde01a49190cd69c6a532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/618edd5e9657dcba2959f42083bbd15836001f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/618edd5e9657dcba2959f42083bbd15836001f87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
