)]}'
{"id":"openstack%2Ftrove~869511","triplet_id":"openstack%2Ftrove~master~I5832733dd312db24d2d8047658fdd1af9f4e700a","project":"openstack/trove","branch":"master","topic":"dbinfo","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-03-21 04:32:32.000000000","reason":"Change was submitted"},"12932":{"account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"last_update":"2024-03-21 04:32:32.000000000","reason":"Change was submitted"},"32324":{"account":{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},"last_update":"2024-03-21 04:32:32.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-08-31 22:55:22.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-01-12 01:21:15.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"33822":{"account":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"last_update":"2024-03-21 04:32:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5832733dd312db24d2d8047658fdd1af9f4e700a","subject":"Establish consistency on db_info.addresses between taskmanager and instance","status":"MERGED","created":"2023-01-06 23:10:48.000000000","updated":"2024-03-21 04:33:38.000000000","submitted":"2024-03-21 04:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869511-dbinfo","meta_rev_id":"89fac14ab7b0498a8179aad30d59bddc8f690ae2","_number":869511,"virtual_id_number":869511,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-21 04:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-21 03:30:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-21 03:30:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-07 00:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-07 01:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 14:40:45.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},"state":"REVIEWER"},{"updated":"2023-01-10 14:40:45.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":33822,"name":"Qiannan Lyu","email":"qiannan.lyu@hotmail.com","username":"qiannan"},"state":"REVIEWER"},{"updated":"2023-01-10 14:40:45.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2023-01-10 14:40:45.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-01-10 14:40:45.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"da5d0f592063f64d4dd8aa21858a6b9b7ec95e36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-06 23:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b9ad1cbbd2d78c1334bc9c00670c4ecfeae90b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:34:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff57e5d2a2cd4185aef6cdc92fe40412\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee8adcf4e3ad4d15b98b71d60bebf6b4 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/493a5311247c4ca1bc6c3bec95b345b1 : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fc79dbf58ad736b1cc56326707f34819fcf4ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:14:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b788de5ebd745b99c904927e622ad69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4011701da264435092eb560dd1853e68 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1b9b5406a446dca607813afe970a5b : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c265a7d27f544ce89c50f84a2bc15d64 : FAILURE in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05c11c0205914013a0a23615ae7e4fd5 : FAILURE in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28cd6cc895f64e0c888c81503caaa25e : SUCCESS in 4m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d9c72c67dcd4ebf8dd247ad8af15726 : SUCCESS in 4m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7841743af86f4cbbb2e4a9950792eff0 : SUCCESS in 1h 42m 25s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/40933d55d4b541839a4b5123d5077ef2 : SUCCESS in 1h 41m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d912a434bd8549a3adde2dad5063f70a : TIMED_OUT in 2h 02m 22s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/46a3bcfaa2a3466bb8f73b1844568807 : FAILURE in 33m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be4dbe699a79cbae9b99b9ad7f80cb3cc7868743","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-07 02:10:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa31a166f1669fc49648c4b9d3a3249b2593f47d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:25:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c4c7dc224548eb80ddbdcc72c2d8fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbc87b0e6325440ca876259414365faf : FAILURE in 8m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af5d23f3bb114fcd96639c658fa21179 : FAILURE in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89f1bbee3ea1b44b6a98150ccbae14fe17fe9377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 04:21:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6de8fea381c04088b05eb62a4e145302\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24925dbd69484b53a0d0935b22ffcea4 : FAILURE in 5m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20770d2adadb4543bdfdf52e5fbb8c00 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcb1ec64a0864fe5a2d227a63cf6441f : FAILURE in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df49e7eb043b40c986e48a71ff34474d : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a391ee45c35e4a2c96417130d326b422 : SUCCESS in 5m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81a81d8c9b0947029d07789f6f2f4164 : SUCCESS in 5m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6e8229cb4f254e56839b6a1d90e6d30a : SUCCESS in 1h 11m 49s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6b8b8ceb22b64ab4ab6d204c364a4995 : SUCCESS in 1h 11m 15s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff54a57858a9415880235e0a44ec3d0e : TIMED_OUT in 2h 02m 22s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3fc8bfca0eed46c58ea05a9e40dab900 : FAILURE in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7dce9f1494d2c7069c1a59dd0057e31de9518d35","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-10 14:19:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec64c9db4cbf1a162350a5c1055b3b39d01588a6","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-10 14:40:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49fd3f7e50721bff5802a5cea7d48181fc65a5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 15:03:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e7ad97dcfe8464c9fafd337b790c8b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a74aad18ef44df7bccff21849fd771f : FAILURE in 6m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83731d3a8c014c219d0962ea89bf39f2 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7c4e83037b84ab5a528164e0f1b2576 : FAILURE in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d514bc7acca4c98a290e3f4be0b550e : FAILURE in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b912c50f214a79a6b88aea23a36334 : SUCCESS in 4m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/018472e5097c414da8d67a635cfb64cc : SUCCESS in 5m 00s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1a6180acb93c4d39b045d6b2f6c4d66a : FAILURE in 30m 11s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5d4585fe27a445de9f3d5d78fb809fa1 : FAILURE in 30m 02s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db1bf33cfb254dd4938c077bbfbf72b4 : FAILURE in 38m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f7d7bc9f7384472facbfb839dc8c0ae7 : FAILURE in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fec90e9d0e5be86af93d8ea18261a6cc3f70bb5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 15:45:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e62c8ab48854c4fb6455054df07001b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8514ed8d7ad84af79f09a20cbffa879f : FAILURE in 8m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a46e5c605eba4e049c92ea01c43eff89 : FAILURE in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3552caf04732b697f70b8cb99f221ef6da0a0363","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-11 14:21:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"070f15ba48fb8ada5cbb48bbfc3e30b27714fd5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 14:34:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f86a9255264030b6e8afc812d07e94\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/031894d8da844805aff25b71a9b27e7c : FAILURE in 9m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5676c8929eb4f5692e797ddf697b87a : FAILURE in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a61d415a056668782f053a96793e15644f109ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:08:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1e5e7d4b94a4400a82a96d76b138d13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22707f3e504d48f3bb55c7fb3044b58b : FAILURE in 6m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174e6d65f2574c96b3e8edb1d5a0f491 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb385f3e1c574d3ba77facca001f1235 : FAILURE in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9b45fcf2e0d4a0491b7680cd7e8ea2b : FAILURE in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508c4a58b44f4454b63e627bc6516606 : SUCCESS in 5m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88262c25f6fc4811b4bac79767ec01f0 : SUCCESS in 5m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1486b46bc29e4c4183e16dfd9eecc6d9 : FAILURE in 29m 48s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0ba06518d2784c5fa1ba905ded98cdf8 : FAILURE in 29m 10s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1963881b41b4a32ab92caf95f9251d1 : FAILURE in 39m 48s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/635715445f5a4ed791e8c85d0ba94208 : FAILURE in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6667c86e5c88d94da12f6a87c0347b7060cfac3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-12 01:21:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60c9317b7079ea554ec598664523a69932df4270","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-20 14:12:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e0b25638bbc4d1fd9af56fae358e297a469ee60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 14:30:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/debf7c26669a421fab7b1e2a8e134de1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a4a80fc37024ce6bd2a722bc63296f4 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f766f148ef124bc4abb1d9bc36a8cc19 : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c517384c6d97475b8fe277f0c4e60cf2570a1da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 16:18:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b635b9caedd849efa2d1a4445855a671\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/950d2caf0aa940afb89805f8b774aa48 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff31daf87e5a42758ae0a10afb5d8a2f : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c47ff639da2b4307b16d6d728417c4aa : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/362f8aa3989e40cc830f6dd896443a4d : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16a7f4a6bd441f191543f24b36fdf1a : SUCCESS in 6m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/381880aca0fd48e3bfb0a96da8930e1d : SUCCESS in 5m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/da03238bdcdb46838bfccb36d830809b : SUCCESS in 1h 11m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/9febe901fbc84621b41070817d6e9147 : SUCCESS in 1h 12m 46s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ef985039c6b413b8aa5c9fc200cb8c5 : TIMED_OUT in 2h 02m 30s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2445991854424cb4b2f175d2f663d416 : FAILURE in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32385cb461ee1a4d07b5fa8f9a36a8c293af900f","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-24 15:36:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"337c8d7b2c67b08e0682b5c7922c84346b0cb098","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-05-26 02:26:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1faa0a1e651bd10b8fcd600a877f27bc993f886d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-02 07:07:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea9755b9a3057e40cc87b755c27decf8b83eca95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-08-31 18:03:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a93c31e7e9e382655f89e6ac0a295f83826742c7","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-08-31 18:04:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9b0ac9448ad29fd26da0d8156379188c9390eb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 18:44:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/585605c6c84e4fd28db938ff19c1f5da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/866a989fbc544cdc96e9ba0f0f125448 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6026a2d8c3144f3e802fdc3e65f1160b : SUCCESS in 16m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4a9f154273a47d6ba5bc00030bd5e48 : SUCCESS in 15m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/571a2df64d54422ea47c02457860859e : RETRY_LIMIT in 1m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"60f00e43dfde68616745ce9fa9a8173443d4d95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 20:21:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb63d090a0c462b89cdf7ca27d2025a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/386e0d051591489c8d2cda14b7d782c7 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28998009386f4bf6bd7f510e49f1463a : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65c210f7d8404a46ad0030f05c7a4056 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4abde859523a49f7b4f6193400faceaf : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d917c56758984febabe8145d10b50309 : SUCCESS in 7m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6022c14c68ac4970b9dd1fe8ddede81b : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c498ec65a49847fd9d9e48e3dcc8049c : SUCCESS in 6m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad75c396573141c6a8776562b86519c1 : SUCCESS in 5m 22s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e39d11b440534b3ea2aca85a9075d342 : SUCCESS in 2h 15m 34s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/bda20020f06c462f8bcfef1be10c5346 : SUCCESS in 1h 38m 00s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0def9d1c9406462682a05632c6a44e7f : SUCCESS in 2h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1da2b6cd4bb2b677c8ac3fb5e5807f657b624a4a","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-08-31 22:55:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8d7bce22bfd7fc4af765d248d3a2a05f08856e1e","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-09-01 02:41:27.000000000","message":"Removed Code-Review-1 by \u003cGERRIT_ACCOUNT_31737\u003e\n","accounts_in_message":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"_revision_number":3},{"id":"16c6c38bd846c4e6a69e05b2c10d45978d53c490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-11-29 21:50:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"32c15f59b2008fed39c3770201a9c2876deec7bc","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-11-29 21:50:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aabc8d2a232c189a76bcb8b6ec7f953733120d43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 22:10:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb552bd998047eeb25c23e343969d75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf91005af2074f869541fba0f8573b30 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0447747b38c74cfca4ced78d9e1cc1b4 : SUCCESS in 16m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e438087fc038474886f416118ba78b25 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ea879dbd860e29ee970ea058abed03da66bf7f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:52:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa0e2a4a22e24506846a1e6fa82c2d00\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/814e868b3813466eb2ff6261ffe01dc7 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5156d608a8a42b3b2525077b3be665e : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9479929688d041d48e28de300a334cc6 : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca849cbf55f34477a40f9bd4eb7a1c0b : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f5f8b34c1840b3afdb0023e8e39761 : SUCCESS in 4m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/46e975ca6a1244e687509dc0ccd58655 : SUCCESS in 1h 32m 44s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/39f36180fa774dd1ac5b6215b60a3875 : SUCCESS in 1h 32m 50s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b6c37e34dec4405a8684e21bca8447fa : SUCCESS in 1h 58m 56s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b0d4bec23c5b4895a310a9ae5399b601 : SUCCESS in 1h 23m 28s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/778191c774e540eba8e5fe1e37946477 : SUCCESS in 1h 25m 16s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/00eb68be5225441ba61a5add2ccd959a : SUCCESS in 1h 50m 49s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/114746b7228e425eabc4c1243e0e56e3 : FAILURE in 1h 31m 41s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/57fed39e74084ee69d098b8f6d65650b : SUCCESS in 1h 27m 19s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0cdff92efe0c4b6b9915c830b0398926 : SUCCESS in 1h 52m 13s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eaa6cc384ea24849853e6b4ff8270424 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":4},{"id":"869c4afae19cd208abbc54b069bf342eb4f2b6f1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-30 02:03:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bbac0b4a3256f2e9d3de97ec4f1eb0ae6d979203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-18 13:26:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_12932\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}],"_revision_number":5},{"id":"8de8bdbf0f42d92811202860bde6797e116f7f5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 13:43:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/454f051150f348d5849d37c286d79dfe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e58f663c0ed44f3d8c1d2e34e545bb2c : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8ed1eea48c84d8ab04f122ce33a09e1 : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc6a2279774248588f381d9310b7637e : SUCCESS in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"173e1284881c0d727ac523b2a9c5339a1c444260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 15:31:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60cee230305a4ef2a65892262a5f46d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c65083b59842472fbca3e4edb665ee05 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed49c03e0ac40ebb2568c4ea5d3121e : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b9a9fe1e2c84171b1a1c63512256279 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/359d0d69ca4b4bda800d7d77e7a0d87e : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0370ee63f44b96b0faf039ed3ece5a : SUCCESS in 5m 27s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ce7c5a2915af463e82c5dc6b52ae0cfc : SUCCESS in 1h 00m 36s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3c418946cfd242c59f8d4b4ea83de0f7 : SUCCESS in 2h 03m 03s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5830e06c22364ce1bb7ffa2d7ee157e7 : SUCCESS in 1h 43m 09s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0213f63357ba4c3492c4ee87c8bb9fda : SUCCESS in 57m 12s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/925914ba38234ba4ba0266601b82eb90 : SUCCESS in 2h 02m 56s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/02f6bf4009304b168a2bf62f5dcb289c : SUCCESS in 1h 49m 36s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/992ab4329d7645d3bd8e6a88d677da0d : SUCCESS in 1h 04m 54s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6dddf813455041e998cd84e8b421c0c3 : FAILURE in 1h 18m 57s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f564f2d0e39645ca942c0601ad45fd7a : SUCCESS in 1h 29m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c72fb14232454ca8917c450972131cc4 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":5},{"id":"e4b621730b97f3dd4242bee61a91476c0ebcb1f6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-21 07:23:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a4b1b793bf9a3e5d66c13e36324e77d568094dc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 07:34:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef5a053460e4e209e27dc059c71da44\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c124dd35b4794afaa66cca4876e56e51 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0948cab134a47898073d9e3682a49ae : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd89cce44f4e45909604e9c167925ba0 : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"721cd958b8e991805b649e9abf8f750a98b3a54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 09:56:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6fcfdf5bafe49ff834dbe114bb811ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce7bedb4bf624f7a88dd482f843bc669 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001886a0025a4e3197c89fb4d3f8b6b9 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d68065ebfceb48e0b91bd4b36414f88a : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90eba0b2a03a4c2dbf9bd1a3ef4eacb1 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f13c9413e941c78d1e26ae10af0305 : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6442e5ffb7324f85acb0f575bedc961a : SUCCESS in 56m 14s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f68a79cf8765412dac239b6f880083bd : SUCCESS in 2h 01m 51s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5416ef881a4e4c6a87506750e90608f1 : SUCCESS in 1h 48m 58s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1994e3bcbee941ab81f8cad48a9ce7a8 : SUCCESS in 53m 51s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bb70e41cf3a14f589cc57fd0ca1293cb : SUCCESS in 2h 01m 58s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2438f21588d141eda14d6192a38fe5de : SUCCESS in 1h 46m 54s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/488cb105e36247a893955073d1b6054d : SUCCESS in 57m 51s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0db05a7f8e914d8a990e913d71970001 : SUCCESS in 2h 25m 03s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b9d6e437edd045abb7f5f5264cbd01b2 : FAILURE in 1h 17m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ba2f84c36dd4a84997d4ed29819c044 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":5},{"id":"fdf493c5b3c885a53c63297b75762bfacaf6adda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-10 08:05:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_12932\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}],"_revision_number":6},{"id":"a106833f6f7fc9832f7213a619c43b30a4b633cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 08:12:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c8fadf505b949f5a692b77ee41425a9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b86d7a84797e454681a90b23afefa730 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffc55b0cae78408e8ec4f5f62f0abb7d : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c5e7669612374431941d646be2535aa3 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9b47f686f8e230bdd5f976dbe9ebe7d06883a87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 09:55:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a448680d524ab7999aecfda930295e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5cbec7ab0f74b78985ed03927f6bed4 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095c2f7ae5b042358760c1000513119c : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5735a522714a4cc7adfa2a49a59c0a41 : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be736a49a70340e78cd993b61a1648f2 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c1abcd36cc408ca55117c762ae681d : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/211cf9b63f6f43b7847a2878e6b9834c : SUCCESS in 59m 49s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/67abe9dc24f54605ac6d445b6f65fb0a : SUCCESS in 1h 47m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/05a97b66880e472891f8fcaedc7c0183 : SUCCESS in 1h 43m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/695ec929b62f49fdaa42f4ee65e51eac : SUCCESS in 40m 45s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c7d5954973b540208ff40b8afab137b2 : SUCCESS in 1h 29m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9add3a205813455bb31641f0861e1e96 : SUCCESS in 1h 17m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6eef1dcb623640449277824b6a6d39c0 : SUCCESS in 45m 11s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/5cf25eec052846c8b62bb316230ae2b0 : SUCCESS in 1h 31m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/be5c0243fcd74046884e0741c0d60bfd : SUCCESS in 1h 12m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a873f26abaad4dabbec08c95123b64e1 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":6},{"id":"b556f926477fd54e7c8d170ffc358aab7c40bf5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-20 08:24:38.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_12932\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}],"_revision_number":7},{"id":"e6f242f23ae7e179cf1f4482427a73ff9bac12c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 08:39:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb08a5eaba8c401389bfc51fbf0e4a58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a72347e4d01f432d97c2cb892ae28adb : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/922a519f8c2648f09080903f135113a5 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ba5884f2dbe43fc9911ee720e1cb37c : SUCCESS in 10m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"60e80b94adf0babba47893d9f28eecf8985d04d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 10:56:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae2710361be34f9b9c18dfc39bbd1b41\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/982cff318a8f4ae094b44a0278679f48 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946bafe43339472688cb11edf627f370 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82c2ccdbc59247648e35dba1138e1fac : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c5aff8f5c754e0aa8523b802e33962a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50088fa2e1d477f888362351b4e74b7 : SUCCESS in 4m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d941da594c5b46d2bd416c267b02960b : SUCCESS in 1h 03m 12s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2202f0c84df04345a69ae3ba1f9a6587 : SUCCESS in 1h 42m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c30a91069aaf4407a34df08bb9600fd5 : SUCCESS in 2h 01m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d454e62b37ea4a558496b240f1de9b82 : SUCCESS in 45m 45s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2cecc5c662f946e39c9721af9ffdfb6a : SUCCESS in 2h 28m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a763dabcb3d041d6835caa79f1bb3b73 : SUCCESS in 1h 19m 23s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/88339591a2b144dcade3fd39fcecc590 : SUCCESS in 48m 30s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4eff296466d24dcdb3569f9c6437516e : FAILURE in 1h 19m 06s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4cd31c1ffa72418eb0d3a2c504637fa5 : SUCCESS in 1h 09m 18s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d80f7f7f92c54883a7dfa69acf14d3cb : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":7},{"id":"43ac11bf8efd6b4b1f122fee8bf781a90e8b172b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-21 03:30:04.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"86c7b055f44e50a2b3f976286f34ea97ccce9d3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 03:30:37.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dd94b67fe4c568174d2d19d31f57b59d778b6ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 04:32:32.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8991962fde444ecd8c64c09b79123a76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d09e98af3384565a806792d9e58d789 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17f9f65d323242c2a4d5b223703c5407 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d8a167e2aa14809b42268956113b0ed : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6080c7f7b45e4710b9ce8c758320fc34 : SUCCESS in 5m 07s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/132ced3e15d54fe1b1c621e9bdeac05f : SUCCESS in 1h 00m 30s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/407447f3406443f7b0349d8c04f73539 : SUCCESS in 55m 54s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e1f60dbf79f94327851e6f704284f85f : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":7},{"id":"69b6869fe263fc9360eec7e18d21f00ee194c7cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 04:32:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"89fac14ab7b0498a8179aad30d59bddc8f690ae2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 04:33:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c618d84d85242fa87432851efdd6168\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf18d35eb014dfc85b408b2e603995c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"86a870cc5043ccebbae54e16d4bb4c3cfb44ee4b","revisions":{"1c012d88f4b359a764824d89afd249d34b37a1e4":{"kind":"REWORK","_number":1,"created":"2023-01-06 23:10:48.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/11/869511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/1"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 23:10:47.000000000","tz":-360},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c012d88f4b359a764824d89afd249d34b37a1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c012d88f4b359a764824d89afd249d34b37a1e4"}]},"branch":"refs/heads/master"},"81978b36c8767ed796c6891d9cee71b452fd1054":{"kind":"REWORK","_number":2,"created":"2023-01-07 02:10:55.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/11/869511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/2"}}},"commit":{"parents":[{"commit":"1d8231d31ffde24387fcb63faa55689bbb81ad36","subject":"Merge \"Fix the missing log request-id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d8231d31ffde24387fcb63faa55689bbb81ad36"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-07 02:09:39.000000000","tz":-360},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81978b36c8767ed796c6891d9cee71b452fd1054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81978b36c8767ed796c6891d9cee71b452fd1054"}]},"branch":"refs/heads/master"},"c81bb4424bb36b6a0f97b396b235c63a7dd1e6db":{"kind":"REWORK","_number":3,"created":"2023-08-31 18:03:56.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/11/869511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/3"}}},"commit":{"parents":[{"commit":"94ec942eaab0496e5ffbf448fa7f1ad2bff7888d","subject":"Merge \"Adds postgres guestagent test driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94ec942eaab0496e5ffbf448fa7f1ad2bff7888d"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-08-31 18:03:50.000000000","tz":-300},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c81bb4424bb36b6a0f97b396b235c63a7dd1e6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c81bb4424bb36b6a0f97b396b235c63a7dd1e6db"}]},"branch":"refs/heads/master"},"362f510aa635fc6c4a21233f41d6359455991f5c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-29 21:50:44.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/11/869511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/4"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"andrewbogott@gmail.com","date":"2023-11-29 21:50:44.000000000","tz":0},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/362f510aa635fc6c4a21233f41d6359455991f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/362f510aa635fc6c4a21233f41d6359455991f5c"}]},"branch":"refs/heads/master","description":"Rebase"},"237e977135d80580fd36078a0a5b7fb02478b906":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-18 13:26:57.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/11/869511/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/5"}}},"commit":{"parents":[{"commit":"8b6ff821a13888336c141f9b006e648bf75a1744","subject":"Cinder Support For Boot Volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b6ff821a13888336c141f9b006e648bf75a1744"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"andrewbogott@gmail.com","date":"2023-12-18 13:26:57.000000000","tz":0},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/237e977135d80580fd36078a0a5b7fb02478b906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/237e977135d80580fd36078a0a5b7fb02478b906"}]},"branch":"refs/heads/master","description":"Rebase"},"1a21f52c688f2b779a05c5be00b2cdc54ddba7c2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-01-10 08:05:51.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/11/869511/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/6"}}},"commit":{"parents":[{"commit":"a427333e809a527ba1a8533caee0f1747158ae81","subject":"[CI] fix coverage name in git ignore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a427333e809a527ba1a8533caee0f1747158ae81"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"andrewbogott@gmail.com","date":"2024-01-10 08:05:51.000000000","tz":0},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a21f52c688f2b779a05c5be00b2cdc54ddba7c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a21f52c688f2b779a05c5be00b2cdc54ddba7c2"}]},"branch":"refs/heads/master","description":"Rebase"},"86a870cc5043ccebbae54e16d4bb4c3cfb44ee4b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-03-20 08:24:38.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/11/869511/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/869511/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/869511/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/869511/7"}}},"commit":{"parents":[{"commit":"6822a76ad9fad71be7875899a6a879439547f523","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6822a76ad9fad71be7875899a6a879439547f523"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-06 20:46:25.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"andrewbogott@gmail.com","date":"2024-03-20 08:24:38.000000000","tz":0},"subject":"Establish consistency on db_info.addresses between taskmanager and instance","message":"Establish consistency on db_info.addresses between taskmanager and instance\n\nWe had two different workflows here using the same variable but expecting\nit to have a totally different format. The taskmanager dns code\nwas populating that variable with nova api output; the database\nlist code populated it manually with a custom-format structure.\n\nThis seems to have survived a long time because these are both edge\ncases: the dns workflow was only traversed with dns integration\nis switched on (unusual) and the database list only relied on IP\naddresses when hostnames weren\u0027t present.\n\nThis patch picks a winner (the custom structure used by the database\nlisting code) and modifies the dns workflow to live with that same format.\n\nThis is spackle over a whole lot of bitrot, but should get\nboth use cases working properly.\n\nStory: #2010077\nTask: #45568\nChange-Id: I5832733dd312db24d2d8047658fdd1af9f4e700a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86a870cc5043ccebbae54e16d4bb4c3cfb44ee4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86a870cc5043ccebbae54e16d4bb4c3cfb44ee4b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
