)]}'
{"id":"openstack%2Ftrove~871524","triplet_id":"openstack%2Ftrove~master~Ib1fcc64d5f27bf595d4b4c1d4e145136160bf842","project":"openstack/trove","branch":"master","topic":"story/2010543","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-03-05 16:21:45.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-02-02 01:22:31.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ib1fcc64d5f27bf595d4b4c1d4e145136160bf842","subject":"Creates the mysqld extra configuration dir","status":"MERGED","created":"2023-01-23 17:39:34.000000000","updated":"2023-03-24 15:54:36.000000000","submitted":"2023-03-05 16:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871524-story/2010543","meta_rev_id":"f0ce3a9887477a109900c27c87095a71ad714c45","_number":871524,"virtual_id_number":871524,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-05 16:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-05 14:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-05 14:23:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 18:14:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-23 18:15:52.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-01-23 19:47:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45eaceb892decfd997ddd3ed08edd5ad34391715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-23 17:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"663f5ec1522f70a689608bae60ff78ae05d31b61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 18:14:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9aab48c70441e48670455c6567df84\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e858630324842eeab6f9a95dc4a6fc8 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/392de46857a34025927b72e38078ffbc : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d6637c8554a72ea97b16e0f46ebf2aaba8671d7","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-23 18:15:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52fd11b258e4db35a3f6fc6a45481ccaeaf9ef28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:47:52.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39414e0d39dd49e98aeed9f729802582\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbe9b2647f6447aa8c684b2e96a916d7 : FAILURE in 5m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d116a78746a43239155fd4114606cbd : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/462901718efa43f09b856dee52a021f1 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7d9eadc95a547cfb50a9bb9804ced5a : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ee0de94707479c804f93a05ad1e9ab : SUCCESS in 4m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90557cad40934a6a97bc06c891053fc7 : SUCCESS in 4m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e3483af4bfc14e9591c5e5b2ab85d1b5 : SUCCESS in 1h 21m 25s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/99e9c14845294f18b353a52c11dd706d : SUCCESS in 1h 12m 37s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cb7dc8af815469d8478865dd8a39372 : TIMED_OUT in 2h 02m 28s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e2baafa98a1243309a4dcb57449ab7a8 : FAILURE in 35m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5836f2033005c8dff12aa9212bd5a9b68b97d2bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-24 02:51:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f19805538621fcea531eac875b3303beb7314cce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 03:20:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d62d322b9d44b35b3dde273871553a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab8705991acd4ad9b40d39d096270466 : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9f7b6eee9d5471c8becbf09c817a429 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1954ecbea446b9225b06b625077774941a298b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 05:02:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcf5797b4ca4bb7a4513e4d4619ef13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a53148f27e7442d6812984f5f1a6d777 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccc8b3c48664ac69ed3400a5b7649d5 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc11c711c4b54880902ad866c01a0a0d : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb5bf5601426437da3f1368c2d479889 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dbb270d0b954efcb4f4d236df5b1df8 : SUCCESS in 5m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d18b96afec924f9f9b36da76ae6d3487 : SUCCESS in 4m 44s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/da25430191ae44adb46cb60a99135550 : SUCCESS in 1h 12m 49s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/43123baf73d9448bbde565b3e1b7c0a3 : SUCCESS in 1h 11m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b339ec1496f42a0823432170ee246ed : TIMED_OUT in 2h 02m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/45e97c0e93da4c9b9828cd5de2899ac3 : FAILURE in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5b824fee80f5ffe03d70985183951a9f9012aa7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-02 01:22:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"366a1ba1685899a288be4dde9d087cd83f1ec5d7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-05 14:23:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c3f04ffef83a75db0c0a85d9ced0e5786d5ecc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 14:23:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c2e9bd2098b8d061c0adeb234552ef517281da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 16:21:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c004810194e419b97b274a846775276\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c888deb334794e1897a840349c95534b : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f33ccee4dcc74a358f16968eb7579a4f : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a625b4d4710a418aa30b95d02d5fc8f8 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6208fb839284c91968e5fdd2f27ff74 : SUCCESS in 4m 35s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2a95fccb06fc426d9a8a6ded3f5f5072 : SUCCESS in 1h 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12d21335648766a6213ce785038dc7ffd81851bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 16:21:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"930cfca9521b8468bef2d7abee6f65efb2314a1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 16:22:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77abdc911ac842de8db2d2da680becd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f937865d4674106b84e037f988421f8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2},{"id":"f0ce3a9887477a109900c27c87095a71ad714c45","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-03-24 15:54:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3de6c1b2aad889e239037bca28539f68f03339d5","revisions":{"40bad5da3db5aa78379dad0e506170dad8c3813c":{"kind":"REWORK","_number":1,"created":"2023-01-23 17:39:34.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/24/871524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/871524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/871524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/871524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/871524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/871524/1"}}},"commit":{"parents":[{"commit":"3789a3951b120ebd0eaaa5a9b196d3cc30418092","subject":"Merge \"add openstack-python3-zed-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3789a3951b120ebd0eaaa5a9b196d3cc30418092"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-23 17:34:29.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-23 17:39:02.000000000","tz":540},"subject":"Creates the mysqld extra configuration dir","message":"Creates the mysqld extra configuration dir\n\nThis PR makes guest-agent create the mysqld extra configuration dir.\nThe original problem is that guest-agent fails to start mysql service\nwith the following error message:\n```\nmysqld: Can\u0027t read dir of \u0027/etc/mysql/mysql.conf.d/\u0027\n\t(Errcode: 2 - No such file or directory)\n```\n\nStory: 2010543\nTask: 47205\nChange-Id: Ib1fcc64d5f27bf595d4b4c1d4e145136160bf842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40bad5da3db5aa78379dad0e506170dad8c3813c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40bad5da3db5aa78379dad0e506170dad8c3813c"}]},"branch":"refs/heads/master"},"3de6c1b2aad889e239037bca28539f68f03339d5":{"kind":"REWORK","_number":2,"created":"2023-01-24 02:51:37.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/24/871524/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/871524/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/871524/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/871524/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/871524/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/871524/2"}}},"commit":{"parents":[{"commit":"7a62e826a402de29b6a929569dab27adb3259975","subject":"Fixing tests with tox 4.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a62e826a402de29b6a929569dab27adb3259975"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-23 17:34:29.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-24 02:20:27.000000000","tz":540},"subject":"Creates the mysqld extra configuration dir","message":"Creates the mysqld extra configuration dir\n\nThis PR makes guest-agent create the mysqld extra configuration dir.\nThe original problem is that guest-agent fails to start mysql service\nwith the following error message:\n```\nmysqld: Can\u0027t read dir of \u0027/etc/mysql/mysql.conf.d/\u0027\n\t(Errcode: 2 - No such file or directory)\n```\n\nStory: 2010543\nTask: 47205\nChange-Id: Ib1fcc64d5f27bf595d4b4c1d4e145136160bf842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3de6c1b2aad889e239037bca28539f68f03339d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3de6c1b2aad889e239037bca28539f68f03339d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
