)]}'
{"id":"openstack%2Ftrove~871951","triplet_id":"openstack%2Ftrove~master~I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b","project":"openstack/trove","branch":"master","topic":"story/2010467","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-03-22 16:28:34.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-01-31 09:27:27.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b","subject":"Fixes permission problem when restoring backup","status":"MERGED","created":"2023-01-27 05:47:30.000000000","updated":"2023-03-22 16:29:33.000000000","submitted":"2023-03-22 16:28:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"871951-story/2010467","meta_rev_id":"b2064ef7a0206955df315af7146a71f186b457fd","_number":871951,"virtual_id_number":871951,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-22 16:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-22 14:41:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-22 14:41:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 06:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 07:59:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 08:24:32.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"57475f160e903cb886f38c4fd1d18ef5e8f17a85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-27 05:47:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f8c0b1c8deda2e4c34dedceb6732754b07f22c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:07:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48b087698de34c1588012452e1b1a18e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89a0d9ca71984143b3cac9d0de455c1c : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31a1dfcf92ed4b0993d3e23d72f572e9 : SUCCESS in 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c0b12e1b34ecffa67aabb966dce61ba1705e4f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-27 06:07:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b5b81cb8d7c0a52acc22f18bbeab2628a2fd282","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6174e0d1af4045e5b47aaf47f75de49f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c62fd4f2d0b4402c8fbeb8393e61176a : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ebf1cede878d4cf08ac1d4b29dbbce97 : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f4ccfa615bace9cb1538c4aa5025f797a776fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 07:59:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0748ab752080432497f6aba74008e938\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dcdd662812f43f0a29025f396a6b147 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f86159fc664686b1d8a1251914d0e5 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bffa4b170b594490af3b77079c500367 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5786fe835154714b335edec3388d19d : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4243f3b36147ab9dfc5ff9782951b9 : SUCCESS in 4m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88669dbd68774c14819ac6af8a94e5a6 : SUCCESS in 6m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8313c06254714bd9a0c0cd4676ae1b1a : SUCCESS in 1h 14m 45s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/589cce9b3031479c975d045f306dc5d6 : SUCCESS in 1h 14m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16f068abbb294ed0815da353d7f04863 : SUCCESS in 1h 49m 46s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4a2c6bf8c02b43e08f8e25d923f89037 : FAILURE in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1dd0360400e8e1152ab7b7c5b3265d6d055244c","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-27 08:24:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c242bcd0a15a2637d8f2d3d8457c042bf9c6578","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-28 06:15:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6814ddcada46469ccd08c570e7c74f493e5da2ad","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-01-31 07:50:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13bc8cf9e841eb3087fc04355aa3352c01b0167e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-31 09:27:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c91b190cacad8cf0e757d0300af9364b81af47a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-01 02:41:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b05e918028f14d93d2d0f96661cfa0faaf7660f4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-01 02:42:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ea40e356dbe6ad617a06b552fac18c4fd8d1ab7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-22 14:41:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef4e434afd3fb3608fc14aa8a6be126ce4a31a78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:42:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dce472aa3744ba77142932aca9d3d9b069809d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 16:28:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a014a7e27a6445928049261bee9bb555\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7436845387984939a030af76e5c0c83f : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90f019032c44c40b39fa148a3836ba5 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78300e8bea764893b2b03e38c5aed89b : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591af76c8d4e49b888a5fb1a352601b4 : SUCCESS in 5m 29s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/73363b99c7bb4f8c80c36f893821e857 : SUCCESS in 1h 45m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba1b368835b41a7955cef8799e7505cec404be94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 16:28:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b2064ef7a0206955df315af7146a71f186b457fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 16:29:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c282513ef13b4dbabfc06e0bf6717642\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d4c385a585431aa5e7f7d02312d63b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"553c578923d30d0eb15ad5dbd0a2c583cae274a4","revisions":{"af9a60d53afc753b9086d02bc1f080ee581bd911":{"kind":"REWORK","_number":1,"created":"2023-01-27 05:47:30.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/51/871951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/871951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/871951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/871951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/871951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/871951/1"}}},"commit":{"parents":[{"commit":"7a62e826a402de29b6a929569dab27adb3259975","subject":"Fixing tests with tox 4.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a62e826a402de29b6a929569dab27adb3259975"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-27 05:34:49.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-27 05:46:36.000000000","tz":540},"subject":"Fixes permission problem when restoring backup","message":"Fixes permission problem when restoring backup\n\nThis PR changes the owner of the file used by the docker container that\nresets the mysql root password.\n\nOriginal problem is the failure to restore mysql:5.7.x backup. The error\noccurs because the container process tries to update the file that owned\nby the others.\n\nStory: 2010467\nTask: 47017\nChange-Id: I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af9a60d53afc753b9086d02bc1f080ee581bd911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af9a60d53afc753b9086d02bc1f080ee581bd911"}]},"branch":"refs/heads/master"},"553c578923d30d0eb15ad5dbd0a2c583cae274a4":{"kind":"REWORK","_number":2,"created":"2023-01-27 06:07:20.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/51/871951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/871951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/871951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/871951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/871951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/871951/2"}}},"commit":{"parents":[{"commit":"7a62e826a402de29b6a929569dab27adb3259975","subject":"Fixing tests with tox 4.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a62e826a402de29b6a929569dab27adb3259975"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-27 05:34:49.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-27 06:06:37.000000000","tz":540},"subject":"Fixes permission problem when restoring backup","message":"Fixes permission problem when restoring backup\n\nThis PR changes the owner of the file used by the docker container that\nresets the mysql root password.\n\nOriginal problem is the failure to restore mysql:5.7.x backup. The error\noccurs because the container process tries to update the file that owned\nby the others.\n\nStory: 2010467\nTask: 47017\nChange-Id: I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/553c578923d30d0eb15ad5dbd0a2c583cae274a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/553c578923d30d0eb15ad5dbd0a2c583cae274a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
