)]}'
{"id":"openstack%2Ftrove~874368","triplet_id":"openstack%2Ftrove~master~I11660751532dde117bbf509302fbde594e3911b1","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-29 04:26:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11660751532dde117bbf509302fbde594e3911b1","subject":"Do not configure removed [DEFAULT] rpc_backend","status":"MERGED","created":"2023-02-20 04:24:54.000000000","updated":"2023-04-29 04:27:55.000000000","submitted":"2023-04-29 04:26:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874368","meta_rev_id":"92f950a2b462e7968fffe3b893f1daf9ba052c53","_number":874368,"virtual_id_number":874368,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-29 04:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-29 03:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-29 03:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 04:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-20 06:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-29 03:20:28.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"5a64d16940756cee3c4fa9017da92d3747328043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 04:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d17ad9ba987d33d5791f01e974fc00dfaeb5037f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 04:38:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd57d19da6554e3b84fd0ac5db8ed491\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c537a86b1d947119fa34380515d6e82 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/634dd49e69f346e69cf5949ddeae5f77 : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"515479ec22e5361da2779967b336ab86099bf82c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 06:33:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8f60b40cda4b16a752461be19ac216\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dcee43068c34ba1ae5b543cd92c023f : FAILURE in 6m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ca8759158143459dc7910407cd5647 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b8ffbfd5e6a48ccbdf19b5046473052 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1d19a3bc16e48c2863566d79198e721 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf42a57f931347cdab292374c32350af : SUCCESS in 5m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/edd59ce28245448da22861b9634ec4d6 : SUCCESS in 4m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b61204e775dd408bb579ead8ca381481 : SUCCESS in 1h 12m 48s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7af7ae2f355d4100b914cc77cdb3048f : SUCCESS in 1h 14m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d00fbe888934af4b72331b7b130e339 : TIMED_OUT in 2h 02m 16s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/690674598c8643a7825b69f2c80775b0 : FAILURE in 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"688ba30b895b73d1a3d1a17631b36570e98ed359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-20 03:29:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2ba674df4c4759b16de61d63a7f2a3c8537b8d88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 03:43:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb78a11db90e4a7da1a2c22ba52684d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41a5668fc7bd492f91c512648846e65a : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28008bb3f8e34f15a1075823fc334540 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a207d98d0d15222dac5dcf708ca432f384c13abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 05:31:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a5b8cff9e46408193e1ff84408f7e68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3a2a28d359f40ae999a8638c8f154f4 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bcee3e3e94546d6b283c3894b158e8a : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124711bf5743467baff166f82e4f811f : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdb359b1896345328c27c1ebe52f6c6b : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b2509e845b46b5a753a0403ca435a0 : SUCCESS in 5m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd0f68ef3dda4c31a4a2fc567dfdbf5d : SUCCESS in 5m 29s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8f951cb2a95543eda3bcdf84cfba00e3 : SUCCESS in 2h 00m 27s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/80ad2e8d3c2348cbb183c2bc16bbf082 : SUCCESS in 1h 42m 37s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/130faf64e6b64562bcb6682bfd8f5e05 : FAILURE in 1h 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6cad7c4e0ab6215705a449e6f8dcbe2e3c2b31c2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-04-29 03:20:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7fcd4bce65aa388dcbc36892ebaac3cc575097dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 03:21:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"726257ac4f715e3dd59e37c6be52711e347ab380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 04:26:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71cbbd975aae4783b31639c948d0bcb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48aaba4048eb4df984f5fdf89ee44da4 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3b0bc32f84e48cdaf423bd7ffb0bb37 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62d15675943d43c4a152a48298f9bfb4 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6a58ae4c8f4367b23582947770813b : SUCCESS in 6m 00s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f0508267dad5415a99b0604e88b898a8 : FAILURE in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ffb61c2e9726a64b31bd839dfc483bd14fddcb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 04:26:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"92f950a2b462e7968fffe3b893f1daf9ba052c53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 04:27:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f08bb1ced7a14f6bbbafafeea0d84883\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c323ae116f104b8db24837ba5f0f0d56 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f62d3197c4be1e78654d9bac3842530b39954b0b","revisions":{"44c2dd44e8896c522ed9a67a0f511ea41cab710b":{"kind":"REWORK","_number":1,"created":"2023-02-20 04:24:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/874368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/874368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/874368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/874368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/874368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/874368/1"}}},"commit":{"parents":[{"commit":"7a62e826a402de29b6a929569dab27adb3259975","subject":"Fixing tests with tox 4.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a62e826a402de29b6a929569dab27adb3259975"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:23:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:24:51.000000000","tz":540},"subject":"Do not configure removed [DEFAULT] rpc_backend","message":"Do not configure removed [DEFAULT] rpc_backend\n\nThis option was removed in oslo.messaging 9.0.0[1].\n\n[1] https://docs.openstack.org/releasenotes/oslo.messaging/stein.html#relnotes-9-0-0-stable-stein-upgrade-notes\n\nChange-Id: I11660751532dde117bbf509302fbde594e3911b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c2dd44e8896c522ed9a67a0f511ea41cab710b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c2dd44e8896c522ed9a67a0f511ea41cab710b"}]},"branch":"refs/heads/master"},"f62d3197c4be1e78654d9bac3842530b39954b0b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-20 03:29:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/68/874368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/874368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/874368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/874368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/874368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/874368/2"}}},"commit":{"parents":[{"commit":"49ac6ae0d29db9686124e5c80b697c7b01ec6ca9","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49ac6ae0d29db9686124e5c80b697c7b01ec6ca9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:23:05.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-03-20 03:29:35.000000000","tz":0},"subject":"Do not configure removed [DEFAULT] rpc_backend","message":"Do not configure removed [DEFAULT] rpc_backend\n\nThis option was removed in oslo.messaging 9.0.0[1].\n\n[1] https://docs.openstack.org/releasenotes/oslo.messaging/stein.html#relnotes-9-0-0-stable-stein-upgrade-notes\n\nChange-Id: I11660751532dde117bbf509302fbde594e3911b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f62d3197c4be1e78654d9bac3842530b39954b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f62d3197c4be1e78654d9bac3842530b39954b0b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
