)]}'
{"id":"openstack%2Ftrove~874370","triplet_id":"openstack%2Ftrove~master~I8f595fff616606a96e70d1712a3a2e65c85b3476","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-28 14:14:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f595fff616606a96e70d1712a3a2e65c85b3476","subject":"container_registry_password should be secret","status":"MERGED","created":"2023-02-20 05:10:22.000000000","updated":"2023-02-28 14:15:36.000000000","submitted":"2023-02-28 14:14:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874370","meta_rev_id":"4ef2182c8f1c49a2d3889831774728ac3c911315","_number":874370,"virtual_id_number":874370,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 14:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":1,"date":"2023-02-20 13:10:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-28 12:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-28 12:26:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 05:18:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-20 07:23:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 13:10:13.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2023-02-28 12:26:26.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"f4f4553b77b71c5a94429700b1ef89827ae6bc06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 05:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af45b27afa689d0cdd6278c4734643b8bf11b000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 05:18:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0518e69cbee4c0a8da725564cbfc197\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7832e329b9624e1d94b7b81109518334 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bedbdb23e0df492c82ffb53f02cd1f95 : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28ca9deebe0d5084c88f6830ad3119e6df49fc40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 07:23:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/090c0050f2084dd99f98c288d4cf6eed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/695a37000e4e4bdda955a0f63a2f3f71 : FAILURE in 5m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744fbb63c68d46798125dce5f2799240 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e4be82adf048e09798c0cc02a0f1a6 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9ee02726a5d4131b655d66edb98f023 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f93ed77fc7428b932b9dcf150bc144 : SUCCESS in 5m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f881cfe80105482da0099445db06b861 : SUCCESS in 5m 18s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f0328e9c700046109799b0a2e6bd091c : SUCCESS in 1h 13m 20s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/3a4eb845cc794ed489ff19a4e55dd87a : SUCCESS in 1h 13m 02s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7c6e097dd9f404e8f5d637fe526f59e : TIMED_OUT in 2h 02m 55s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7529d14ea1834a1baaa29282f18e3a2d : FAILURE in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5154960831188024376da604f660f24965dea93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 07:37:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"569537eb72de81c46a4f045049f12f9becc159d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 07:53:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1582bd2b317242878b170d16c76ffd02\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46af0e52fa734479bb51cd9c304b5648 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4bfb016be8994bde9d40f068d91a5870 : SUCCESS in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50abc40f168cde3456ccebea11c1e014fb514223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 09:42:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4202a49a7d484aa75141b04996c701\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ad1c6180bc945858ef864b8ce11734a : FAILURE in 5m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660fd2446421407dac3c675c7412f30c : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30bc77b9aef2403d95dba856ed9df91f : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ff8f4df6f8f4f44aba98ab84550d90d : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa452b1bfa047e0b5d790c0062e6566 : SUCCESS in 4m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cab9d6c21f644ba84d8bcdf94267090 : SUCCESS in 4m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/13043239525249359c2acb0ef18a5477 : SUCCESS in 1h 38m 23s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7c0fa0abe18c4e93ba5a607160d08558 : SUCCESS in 1h 25m 24s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/574e8d0107b74b89a6449fb75eaccbeb : TIMED_OUT in 2h 03m 15s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6fd2c39a92fd4fb8b72af1f4de2ddd34 : FAILURE in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f200985724f342437247c546c7c03fec33c043a5","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-02-20 13:10:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8d38977d6560509ae3daa6d53067408728fc212","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-28 12:26:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68563d85ce30ff0460ea4cd1558e30670645bb19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:26:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"546932926bb3a1e3793982dce855705b8f687586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:14:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7acc60150b24fc881c01d867008f04a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72905c354b3b4597947428bbdf7f4b60 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3123e2ebf574a21ac9daa747c8d0b2e : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4a0698068bb4491882649c37b479358 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea55c00b47004a1a82086ed8df74f05c : SUCCESS in 5m 46s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0982c32a0c5d4ce4b2a3bc536fbc9edb : FAILURE in 1h 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc8a760fca335496697698b72402e483413f511e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:14:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ef2182c8f1c49a2d3889831774728ac3c911315","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:15:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd576301e8746b59071286e61c7345d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07a820de0004312a08fd1ddf011be3c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f7d01216ad00feda86135d72125640df611f3c0","revisions":{"e5a3f56e6cb475fadf6075fddb547931b5fe90ed":{"kind":"REWORK","_number":1,"created":"2023-02-20 05:10:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/874370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/874370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/874370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/874370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/874370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/874370/1"}}},"commit":{"parents":[{"commit":"7a62e826a402de29b6a929569dab27adb3259975","subject":"Fixing tests with tox 4.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a62e826a402de29b6a929569dab27adb3259975"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 05:08:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 05:09:42.000000000","tz":540},"subject":"container_registry_password should be secret","message":"container_registry_password should be secret\n\nThe [guest_agent] container_registry_password option takes a credential\nto access a container registry, thus its value should be hidden from\ndebug logs.\n\nChange-Id: I8f595fff616606a96e70d1712a3a2e65c85b3476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5a3f56e6cb475fadf6075fddb547931b5fe90ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5a3f56e6cb475fadf6075fddb547931b5fe90ed"}]},"branch":"refs/heads/master"},"8f7d01216ad00feda86135d72125640df611f3c0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-20 07:37:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/874370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/874370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/874370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/874370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/874370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/874370/2"}}},"commit":{"parents":[{"commit":"7a62e826a402de29b6a929569dab27adb3259975","subject":"Fixing tests with tox 4.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a62e826a402de29b6a929569dab27adb3259975"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 05:08:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 07:37:46.000000000","tz":0},"subject":"container_registry_password should be secret","message":"container_registry_password should be secret\n\nThe [guest_agent] container_registry_password option takes a credential\nto access a container registry, thus its value should be hidden from\ndebug logs.\n\nStory: 2010594\nTask: 47456\nChange-Id: I8f595fff616606a96e70d1712a3a2e65c85b3476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f7d01216ad00feda86135d72125640df611f3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f7d01216ad00feda86135d72125640df611f3c0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
