)]}'
{"id":"openstack%2Ftrove~875661","triplet_id":"openstack%2Ftrove~master~Id6cec4d2f94bcdbdf0954f1fe3be1ce4404c983c","project":"openstack/trove","branch":"master","topic":"tox-cover","hashtags":[],"change_id":"Id6cec4d2f94bcdbdf0954f1fe3be1ce4404c983c","subject":"Fix tox cover check","status":"MERGED","created":"2023-02-28 12:48:20.000000000","updated":"2023-03-05 16:45:50.000000000","submitted":"2023-03-05 16:44:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875661-tox-cover","meta_rev_id":"bbbea4b83df57c9163243330cc0a3c31051100ba","_number":875661,"virtual_id_number":875661,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-05 16:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-05 14:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-05 14:23:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 12:55:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 15:07:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb85143d41182075cf4aa12509da3094a7049f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-28 12:48:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a87953a77b3f4c96476ee8bd7f28d50f14bbfe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:55:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41bf8bb4671c486db374af173927ef69\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4405ddf12dc5423a84dc52954bcb7f7c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28f366aa4f5949ec97d9e47062dbc3cb : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e001c80fce866a7eafd059b844ce26c61d7f622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:07:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eec65ea99ab7492bb77c479b73e0c3c4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/469390a2780143ceade859506b0a1869 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09101d10b73546c2a43bb1ae63b303e3 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c65878bad0574b0b8cdcd5b00945d765 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7fd42c4b35c44d08f3aa723d4768f87 : SUCCESS in 7m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/678a4347601346d3ba2f4a3cfc5ec9b1 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57018e0d3a9744aea72a45921c00cb40 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33cbf019bf98404c8cf22249dee5bff2 : SUCCESS in 5m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07a8db326a3f42d4aead522ecb673cc5 : SUCCESS in 4m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5793ef12f0d143dc9d729c89b7cacdc3 : FAILURE in 1h 54m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6a3ccc9d4dc145569a89eccf759f840b : SUCCESS in 1h 41m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f137c4f468b491d85a1ef4a31df8977 : FAILURE in 1h 34m 32s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fbaaf41a74964ab9b403f349a8060345 : FAILURE in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64a3b575fd00775c1946cf3849e3cdbd7447753c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-05 14:23:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66abb181e6f30a2a211b8cc5c18e3febfb7bb04d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 14:24:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b6b49c8380732e15fb9c540a3753d8faa1a852b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 16:44:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6704f241567457ba070fdae2c249af7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a037c9d6693f43019f591b7479804125 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25808c189caf4fd8b9625e9cdf3aa2c3 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2b96fb20c2b4dce81b3a7bc42b752bd : SUCCESS in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7fe2d3fb33c47c695d564d6b98cf466 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657663bc145343ec9b78ee5e61bbb8d5 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30bfa5bebeec4df2ae0cae03a69b0a05 : SUCCESS in 5m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2b60b70a26f8405cbaafec54aaab77aa : FAILURE in 2h 04m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d54f718b5a444e789a8148c8b58236f7124b6a00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 16:44:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bbbea4b83df57c9163243330cc0a3c31051100ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 16:45:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d18da504134647985166c692a1ea77\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4b95c65e42424034b691f6fdb612f7c5 : SUCCESS in 36s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08131ef49304b7294dd43398f653377 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5a195b8340c445c8bd03abd27f3962e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad1e5fe464d5ee22843825f95d46cda34bd09b0a","revisions":{"ad1e5fe464d5ee22843825f95d46cda34bd09b0a":{"kind":"REWORK","_number":1,"created":"2023-02-28 12:48:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/875661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/875661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/875661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/875661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/875661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/875661/1"}}},"commit":{"parents":[{"commit":"60a202d06539ca6da9136b7dfab1275ddade5fb6","subject":"Merge \"Using local registry in devstack.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60a202d06539ca6da9136b7dfab1275ddade5fb6"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-28 12:47:52.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-02-28 12:47:52.000000000","tz":480},"subject":"Fix tox cover check","message":"Fix tox cover check\n\nChange-Id: Id6cec4d2f94bcdbdf0954f1fe3be1ce4404c983c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ad1e5fe464d5ee22843825f95d46cda34bd09b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ad1e5fe464d5ee22843825f95d46cda34bd09b0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
