)]}'
{"id":"openstack%2Ftrove~876526","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I786d258cc42e43ba1e269d44b302d8aa2cb7094b","project":"openstack/trove","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I786d258cc42e43ba1e269d44b302d8aa2cb7094b","subject":"Update .gitreview for stable/2023.1","status":"MERGED","created":"2023-03-06 09:12:18.000000000","updated":"2023-03-06 11:22:16.000000000","submitted":"2023-03-06 11:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876526-create-2023.1","meta_rev_id":"1a82c9655a49cdf7d37934d6247344c617841e18","_number":876526,"virtual_id_number":876526,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-06 11:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-06 10:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-06 10:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 09:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-06 09:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 10:46:35.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"bf0b63527420747df868234ef08f21d74be8d006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-03-06 09:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f60a2b1a0e02fdd83abe53de3ac486a71de97639","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 09:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e24d552e52ef4e36a34cd659cfef129f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fb117561e5e4b629dba9c58c4f199db : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f8f23942c2a4041b1e1548e322cee0e : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01d6ce3cea3b2f709d87f9c427106182684fdc93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 09:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a51e892339d4414197cbafb25bb5aa1c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0974ed95eb2549bd81e847715413716b : FAILURE in 5m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee7e158d96c4a0d915bef45d2e12e66 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df24c2d0b0b64a58bfc4dc339d7f2367 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ffd50217aac4c2bbb3790f80ac34269 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3521caef9a7b401fb65ce50edd3cd6f7 : SUCCESS in 5m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e2d55e1ac467415690995c8f0d673f51 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aab156f5ea1d72bcd808d8c75e50f7a281156eff","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-06 10:46:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81f13407aa2b151b14c4b8b636679072273f681e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 10:46:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eb55a24299de11aa345c5476c910eb2ff5e0dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 11:21:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b82d8dc1f74003ae4d8f892372d029\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c05bf9d8e2c46208a8bd64d38b07ebe : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a45a7f2a2f4f79910d83f0de5cb9ae : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/478e7793370f4770ac0b18d5784f7716 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e533d0b91945dd8224572c53b98c7c : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7fab02f4b81db1a581c1adfa38029e446a3532a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 11:21:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a82c9655a49cdf7d37934d6247344c617841e18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 11:22:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77ad1d1b70ba44f8b2616e18e75bc8d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91934f7c74824a9d8d01a09e385582d7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6728c1d3be4897a561203e219fa46a718eb8def","revisions":{"e6728c1d3be4897a561203e219fa46a718eb8def":{"kind":"REWORK","_number":1,"created":"2023-03-06 09:12:18.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/26/876526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/876526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/876526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/876526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/876526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/876526/1"}}},"commit":{"parents":[{"commit":"2fbbb09d22039f672ef4a084617bff8748fba134","subject":"Merge \"Remove unmaintained zuul checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2fbbb09d22039f672ef4a084617bff8748fba134"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-06 09:12:17.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-06 09:12:17.000000000","tz":0},"subject":"Update .gitreview for stable/2023.1","message":"Update .gitreview for stable/2023.1\n\nChange-Id: I786d258cc42e43ba1e269d44b302d8aa2cb7094b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e6728c1d3be4897a561203e219fa46a718eb8def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e6728c1d3be4897a561203e219fa46a718eb8def"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[]}
