)]}'
{"id":"openstack%2Ftrove~877071","triplet_id":"openstack%2Ftrove~master~I4c000363dd046fc82f099de0a30c16b1f01d6d2c","project":"openstack/trove","branch":"master","topic":"postgresql-actions","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-08-16 02:08:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28691\u003e replied on the change","reason_account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-12-25 11:20:56.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2024-12-25 13:52:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c000363dd046fc82f099de0a30c16b1f01d6d2c","subject":"Working on missing postgresql features","status":"MERGED","created":"2023-03-10 06:53:06.000000000","updated":"2024-12-25 13:53:40.000000000","submitted":"2024-12-25 13:52:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"877071-postgresql-actions","meta_rev_id":"b008dd97fca0bb9dffd75a708dc1efb791d30f8b","_number":877071,"virtual_id_number":877071,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-25 13:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":1,"date":"2024-12-25 05:25:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-24 11:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-24 11:59:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 07:05:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-10 07:50:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 13:52:55.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-07-20 13:04:52.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-08-16 02:08:09.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2024-09-27 05:49:54.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"}],"messages":[{"id":"7519b264de54eaf2d440f6211c5ff66b0da033ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-10 06:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e40e717cee38a725fb24655362a7234d8a717ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-10 06:53:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"91793cd08501781cf1fcc7e3a4643907f9091ef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 07:05:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b010bac52fcc4f1ea0c6893aa7208c60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e85c55834b31472e924e004243c09901 : FAILURE in 9m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6687204b3a2046869c21c91841ae22da : FAILURE in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30ff94bd7cf3be8a9ed3c48f7ae930ac366ef23e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 07:50:32.000000000","message":"Patch Set 2: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e4880e4c290457d978198c2315ab8af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0018917a8e44c15bd8991932f8ec68b : FAILURE in 5m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64bb3e13a9a49b8847ec6e2ae94b353 : FAILURE in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75f31a62017e4533bb08c82c8cb45210 : FAILURE in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0783a75ddbf74ea29542d5ac111b0ecf : FAILURE in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7bc2f90f02043e7b57d0d88bdab0278 : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3a5538beda04a4b9291d1e801f41e3d : SUCCESS in 5m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f55d7983213d4cf4b1b2b4fc15ca55f1 : FAILURE in 33m 28s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1ea6ffd2648848509231a386b026f8c6 : FAILURE in 33m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a80eeeb5bc854f278b17d57bde948a5b : FAILURE in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f93d45adc869ba5bb6131bd7d0cfdafc0b951429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-20 11:06:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21ebc0ca94b8173961a7fdf2fc1acc888ab0ac42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-20 11:06:59.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"cbbbbd300b029106a565a442506892ee8bcb1ca3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:23:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7701db8977eb4a508048f9107c167a6e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4718716a7eb4a508ecf3314c83d84e2 : FAILURE in 16m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edad29a00fa24734b2b83bef39705e7e : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb678340795e17e1b085e7d0eb42b514d79d33b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:51:25.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b57dd8dd1d1a4892b722db6b7563a69d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab93e0f13a0445ee9707695d8ad525c6 : FAILURE in 6m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e26c93c7904bdf83a491b6431e97ef : FAILURE in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f977c8078aa48bcb0e5db161a0da1f4 : FAILURE in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/082504f5eb91481ea62c1d0f69684666 : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d24c534d7dbf433e9d8dbf2400fe3c22 : SUCCESS in 5m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2fa200145ba141eb929f0efcc3e1506c : SUCCESS in 5m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bec67bfc455a4d009fb39b13ae26bfcd : FAILURE in 33m 55s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/63cf5b9dced54608bd44c9a2c5889836 : FAILURE in 32m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f297416641f34279a1ea1ed474ebe06c : FAILURE in 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d4b71a457ef62afd93bfe94bd0dcbfe670611db2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-20 12:56:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1333a33a0a93fc27a9bd4773f3f630421966b225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-20 12:57:31.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"ce5c0c99d72b3f8fbb14d49420ad79b7896a3c49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 13:12:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7890576158eb43d48d08b650740c81e3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab6b731476e34243b87bb825031a389c : FAILURE in 13m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/642944544c2d4e0d9b226fc64a80297e : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bd62d7fb5c4a0fde68b2e4368779ec49057e95b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 13:33:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3201e50322ed4e20b36da73865118201\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/648fde72bff9442ea480bc7037d80670 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535f63693723491fa9d06e0b7f52a503 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef4dbda189ee48809da6b32b65c5fcfe : FAILURE in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d451ff3fb024659bd3b67d300fbfd1a : FAILURE in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a317ca0efbd4e5288b89b0c04f138cb : SUCCESS in 6m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b8444e0a9f5418589af61a041584a7a : SUCCESS in 5m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/048bf3d5edca4666a9a12a561d5ba974 : FAILURE in 33m 29s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0d373165dd414381985b4a3b0a0eca08 : FAILURE in 33m 51s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6383586c6c05403da0a77dbf59e30fce : FAILURE in 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"627f3454d830bbce5fa474b0fe17e6f94d3406ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-21 13:20:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f8914d882e7c2fbd56be2ae7aa2200dc6703e8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-21 13:20:58.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"a0f7995ad9550ee73ccf093baf1213a595675925","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 13:50:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5d22ba38d3400590f7b3eb141e16b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0234294431ac41df845a9e0d7a0b4122 : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b1b3b391c6e4bcaa937f24f6f21dd9f : SUCCESS in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2d6239c2a7d32293c1f66638709e36c200547691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:50:22.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dd29bf7bb404a30bc1738c87dac4436\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d84abb877c447caeed9bdec7e0ea65 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb827d1083a4438fb7ecaf155cdeb445 : FAILURE in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97a28255d0d54f1887b591d57456d298 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc14a3fd7c7846febf8e9750c77a68e4 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a30a78caf0b47f2ae7ac385e2f5f5df : SUCCESS in 6m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cec68eb982214e06b245000f3bef45c9 : SUCCESS in 4m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7ef41533bf9949489b3d86b41decc6fa : FAILURE in 1h 29m 08s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/10c2bd2575c94420ab34aec28fde35ad : FAILURE in 1h 55m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01ab43157ade40ccb2d116d16cc512bb : FAILURE in 1h 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8f6f4fa594a0dbba5f68904c838930311f387378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-22 06:46:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d402435d75b2f9093d82d0b2d0669434f68eaa2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-03-22 06:46:42.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"ac87f8b2b6529c3528ba0534e48d93af2329c0d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 07:01:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581530dc9c024d8dad13baadbb352ac4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f4f0db7be704f769bfeb7619cde7def : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8ee3c5348014e68b24df99abc922e9f : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3e432f061bba79b31f6a1a8ac456472e9b5ba1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 08:56:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62a972047440406f8797b4a05bc28b57\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15f2730ba5474d17acef101f43c0bd20 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd472f5750bc40af82850100f48fcfd3 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d6ea84450c54720959e1697b3e50f94 : SUCCESS in 8m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1207d14462c54df49bfed8b078b5b851 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87be902d0d654130be657d258f2fd942 : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93f7e15f1e5a47ccb971dabc6771612f : SUCCESS in 5m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c4c05167a7234272922d93cf78b74023 : FAILURE in 2h 07m 53s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ff5d6642dc4547c29d34545ba4270ccd : SUCCESS in 1h 18m 37s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/411679786a544406baa98cfe16a69673 : FAILURE in 1h 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f61248c205e344c2c7033d9cab576cc50380f704","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-04-03 05:59:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ce17fd45deac86ea6f793ea1f12e2b257a92e329","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-01 13:52:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"82add08e6d73d246687882a61a8b3a42cafbdce2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 14:09:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c7356aa64f6411f9244649c49d29ebd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e0e2a78d58c4042acdf35072d170194 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/597965351f8041ee8386ed29d5bd517f : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bdfb06f60234e9288831b1569aac64c : SUCCESS in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b440d29896ffdbec218380e2ebc01ef992a5b896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:11:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5a3d04b83a49e4ad4ea31c69eed03c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5277482b4f2e4c658042f29a0dcb4d84 : FAILURE in 7m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d381db7bf66e436da9a26013158a3168 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a72bf0e7621a4be484c79dcfd740a3e4 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f89c0b1b0fa741baa3997fc10a686f92 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bfb6dd621364e209983911be62291a4 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725b690c7cfe40daafa8832994b19821 : SUCCESS in 5m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/70a27864a25344d2beb072986cb8d0ab : SUCCESS in 6m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f0c7ff0ce851432cb5c8458be9221454 : FAILURE in 1h 15m 03s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2bf01e98da4e422996c42f92a2f69a63 : SUCCESS in 1h 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"67fefc01a564d451547fe16cf1ff0f6fc84b843b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-20 03:18:03.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"92531249c0270f8773535f458ae4e138347d8ed8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 03:26:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161a5b2f8f334dbcb53206713184682f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ef3fde0d4b54bb688b2c18dadcf9579 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ef4482011af45939180e6e70ca0eb0f : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7d1e1e58ffc47beafd06d68c52fa05b : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d02aa6032c93e2dd935ed724f2d6c47d9e00b1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 05:30:34.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e412e3fe65124179ac78a6bd1eebd181\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b1dd9cc129448a09d85a739114aca66 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/195bd382fc5a4f1cbb5f78c39313cea4 : FAILURE in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34068a0667eb441ca7c0bc1d68647cf8 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b6ad40702754436bb682fb495cb59e3 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9be4aee62ab34e679c2c893490f0d544 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd82dc79cbd54c11a9977455caed73ae : SUCCESS in 8m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2d6ac2556bf54e5ba980dee15e66d7de : SUCCESS in 5m 40s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/60ae1c9024cc4675b2d2a57f37a5f792 : FAILURE in 2h 11m 32s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/707b506f1c2d4210b45ac1698b10d9f5 : SUCCESS in 2h 09m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c768943345a881e0e7fb2c3baaecbb6c095ac1d3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-20 05:47:53.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c3546a9554cfa98732ed69dea5cf15af29ac7ece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-07-20 06:57:28.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e071129cad733f47264ee43a0bf88650e074c7cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 07:12:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f896f2082245aa92c1212d57d3e8eb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0996f69ba6654d0d981847320a64a28d : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4764f33072f54d89a26a06048703e896 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75d2f1d2c0ed44cf9421058d1df05047 : SUCCESS in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ca8a01d04af8e4dec8bb49cfae45b25fc78c9472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-07-20 08:04:59.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"9f115d724e44158da813fa5d98cb947f06998c90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 08:20:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9864fdecdd8405fafaf53aac5d9e89b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9265b1793db41a9980c4c7b0278d372 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a3cf0391e694665bb322b2db6f70564 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b781c054f7c54037a858d4f521eadc0e : SUCCESS in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fb943d2cb953d7097379119a47cd0657cc00e271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:13:24.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd28d3de41754225a8c963fe90306b30\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95249f66d2654d5f8d06fe8a1ac6a2d4 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a2911562f3d4a95b3b5a0bf5d1c918c : FAILURE in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5aba630c45d419b9b13276d83edda62 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12881898fef441acbda17d3fd0cf7562 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88c9c04e7be34637acb31ee8e3b67e30 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5ce47d4d384d079a36c4483159e5b1 : SUCCESS in 5m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/174d2f6058f44360a852bf0f3b61c7ac : SUCCESS in 7m 41s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e290f01da80a4a5ca04694a959c8f723 : TIMED_OUT in 3h 04m 01s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/57ac6d3dba9f4cc7ab6735381e1ba80a : FAILURE in 2h 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"098c70ea86f3713f78a86db31c9c7ff62140c596","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-20 13:04:43.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7f375aec2f99fab67975bc1a982f99f8b79e372f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-20 13:04:52.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"840206885089a74d471160ae82c122b8bd36f699","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-07-21 07:29:39.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"dc06d3de75d1ce29bb4fbda3d51a37cce501af18","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-21 07:35:24.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5e4b71c9e06f428dcac52242494eefb2424d6eb5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-07 09:24:30.000000000","message":"Patch Set 13: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e856904a0ee5dccff0417f9058efef042e6320bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 03:18:11.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_34671\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"_revision_number":14},{"id":"eb25654e3197fef7497a854ee36de2329d6eb08b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 03:22:01.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7aa345b8b36c5a24c4788cb485c17c9fb91d76da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 03:34:21.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a4d8a010a974e12af25b18292f31e57\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72532031770649399ce129041ac35979 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ce104e6ecbf47829f7bb9bf74c2914d : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0fe10e1b665b5973f252830543e74bcf1ce1c202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:31:35.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0e7868351724e45b87a47f1dad45cb7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/365ac61fba58496497acb11299aef528 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd0931969bc4a88b8c3ae65cf2fbab2 : FAILURE in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f06a19aa70734dc6a0a344aa1788bc4c : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29a49e666043431ea630c8ac8dbe3285 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e31c23c16e4941f19d21a74afc2a5239 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cba99190219465abae76e00e166f7c5 : SUCCESS in 5m 06s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9f41daeb41094064949971935d504d03 : FAILURE in 48m 35s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf585f09dc6c41ba8189acc190a52a9e : FAILURE in 29m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f73717f26ae240e9ba04a72ea6e4c2aa : FAILURE in 30m 22s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/232ee089fd6640a88ce89d46e3494ccb : FAILURE in 54m 01s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/50496525e0f84a7cab043b8153b22a69 : FAILURE in 30m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/01f55f965e8f486895f72597ac8d9196 : FAILURE in 29m 10s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a05e0e2e47c440b98c000b5e5d10aa88 : SUCCESS in 58m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f88a09a20f5e4551bf6c972e7f63af94 : FAILURE in 32m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6677aaa64c364c06b5a17d7e511f9b3c : FAILURE in 43m 35s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/729e7a9946854f79b1287abd30d973eb : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":14},{"id":"3bf456073208d5c054bb70e8ff6dbbe25a663754","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-04-02 06:38:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"eb96c059163025cfa1ab7b64ba8644ca3e5bf047","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-04-02 07:01:32.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"89ef8412a0c0457ed7df32bbfdc0b3da3aa31aba","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-04-02 07:01:58.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a6bfc326c5a2b3c051116d1a103b3381fb5ed98b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 07:13:20.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c452e8639e4d6a879c3c91da67c42a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13eceb5e45eb46e280b594817a5d61ba : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bc6179a6ef24e6ca682704bd0536a53 : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"57e96923e49f73bc0dc294d99fb191189ba9b461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 09:08:29.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d45c0f5b76b84491b93ceee9086bd72f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/611597b6ecf9460f9489a2431e2d86ed : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31971e77b6f44d11a164bbb9e51bd192 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1de0ed5bc5e4236ad36300a252454ea : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/068122ddd20b4901baa0e6478fd16faa : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e24ecd41ac954bc0be305f114734ff39 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a60c8ae83d4a8983a7633ca6ca5655 : SUCCESS in 4m 51s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/cfd1abeac28e43bc9c56b1c2c8cf3ba1 : FAILURE in 49m 22s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/456678c2b4994ad2b6de205d143255ef : FAILURE in 28m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4611e1c57b0b438ab93ac59cdacc2984 : SUCCESS in 1h 51m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0bcac20f8aa347e498af08ae475dffa1 : FAILURE in 46m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/36bb769eafc1445099326a5afd8a88e9 : FAILURE in 20m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a2dd4ae5bd644cefad8cd46b6e9e5677 : FAILURE in 21m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/04247311afbb44c49396eea6b1afef34 : SUCCESS in 54m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e0f5d43eaada4489849eca4274f73945 : FAILURE in 25m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e0b6b0619bba4fc7b44126d2d32a9838 : FAILURE in 1h 01m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a39ba028ba14f429e8df8fc03baf39f : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":15},{"id":"e5a22c24b65389e59af84b52ab95628228a4dd7c","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-04-02 10:16:35.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"65fea65fdd05958737ab488e7fb54bb9dd4519d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 10:33:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da4b7046e66a46c1bf49e3c4ec3ac885\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a81cb960939d449a91afa3d95f190c1b : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21ae1c3106994739979c8b13908cb6f6 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f1863945cd86023be0b444114c96c7633ce156cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-04-02 10:59:00.000000000","message":"Patch Set 16: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"b88e42bd65212f208dccc0e8e82fc7934ea557d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 11:09:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a64c82f5a2476d97d11ecbd4f0b2d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ad4e9329c1744989a8f9001decffac1 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/81cf9fa2986c489aa30b9e7e5e86b728 : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0aa744cf6189988805c8c17f2be167710464e331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 12:55:47.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ed41d8e84344464860b4bc6dc6a10a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b81d2f20081544658a0fa4f414398f9b : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6c499c74154ff08051644d259cb2a7 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a13221b1f5a45a183bb404d8c264d0d : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e64a7a8367114f5997b305479a255731 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0e1e68dfae1495a8c896093daab15eb : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa5d3ba817a44a5971d92d6237b5fab : SUCCESS in 5m 05s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/1e58e25d368749e7bbf89a1f43e433ec : FAILURE in 47m 01s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a8e517476aca41009c067544c138d5cf : SUCCESS in 1h 38m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/62871013edf947b3a24c7d4c69d2f398 : SUCCESS in 1h 48m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/763c941582b147d697baf2f9aede093e : FAILURE in 44m 00s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8816b0ffa9154b12aa36455fe76c8084 : SUCCESS in 1h 55m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9e09db664c5e4e2cbafc148d82efdc1d : SUCCESS in 1h 18m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/dbf6a0dab2b04e8380e4a8504dd9f347 : SUCCESS in 48m 13s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b8e19a5f6a624ad2b4a3cdf360cc12cf : SUCCESS in 1h 29m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/aad611b5fe1e41ac8be4badb5d1c7d3b : SUCCESS in 1h 18m 07s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84a4152e1fb248f2a2fb01094bbf5969 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":16},{"id":"63be806921d6003bc05f97c0fd8bd0cc910fb204","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:25:03.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4bbe71530398c87745e83b9ac3c0336124ba8d06","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-28 01:06:49.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"8f53d2ecb91212596251c9a8acf2a6d5876dda30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 01:19:47.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a7b175c7844d57ade81144d6affbf0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d7834e7f93f41fa8c501b591401c58b : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35aaeb50f54f4004a4ead93be67eeb27 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"993eefbf108ad11ef2bf835ca98a61a411447444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 03:14:27.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d486abdb9f147cca79e758438a97958\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c77a1f127fe4e308e65ce3deb010b67 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c60c053ca04d6b842269f5722f2b16 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3491854980b5447696d08acea558a3a8 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed8037cd0890408fb8ed7c023822b74b : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d7084293518488c8431f33fffa2a6fe : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef23432375140a0b15544c96c74aa98 : SUCCESS in 4m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b9121277e0cd485b893c09804246416e : FAILURE in 45m 38s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/215d4b5c42a94b2b99113a6d58486655 : FAILURE in 1h 25m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/9374f6cf82d64bceaa108fd372858cd8 : FAILURE in 1h 05m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/714e9ac3ef6341e3920328290b43ad6a : FAILURE in 54m 11s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/df2e94e1ed2a48009685140fd004baef : SUCCESS in 1h 58m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c58cd027e3474ea8a1cd5932fe169dbe : FAILURE in 37m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/f4dab61d0a8147f9978c89e04245e670 : SUCCESS in 48m 57s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/827a2488b48b4b44a6c055cfc300e6bf : FAILURE in 1h 17m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/01d16e94a9f3447da964a7ae69e45e28 : FAILURE in 1h 59m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/080e7ee812f546369ddb2b14d94dbc8f : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":16},{"id":"72c17acbbb953b29a61df7d1ef66add763026341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-05-17 08:13:08.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"8b73edf2be03babde2a11f92f85ba68dbbcb47a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-05-17 08:13:28.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"7ceea224fb4ee0c223582ed0e9aa62ca67565799","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-05-17 08:14:24.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"0fd4ae822dc6ba105593e8eaf27f3f06dd127d13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 08:25:24.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5127844785214a50a60e999144a22bdf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6156b1ae1a4a4d34bf43af25f6f9f6f0 : FAILURE in 10m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2fecd29febae48b5bea1b450640ed789 : FAILURE in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c5b325c28cb451354440fe63c7d28321ab92d68c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 08:36:40.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c74302c4598415c8c0e0d43adba5f70\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/382f7bcd2a294f2d97490d215b4ee6d9 : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e1f66db46c41968bb3be801f2abcab : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc675f425a114beea4cc228b14f4603f : FAILURE in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7df676f7ff7c441086c45cfc3b302ad4 : FAILURE in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73810cae84d54860a07e274abf2241d4 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e715f8d6c00f4b44b95b9c11898bd9ea : SUCCESS in 4m 32s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/5abc630535664ff6a5a5c175913f5868 : FAILURE in 18m 16s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/254af65956094d9eb405c6a8dec9a222 : FAILURE in 20m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/ea095c023f94403798c7b5d2ac2eeceb : FAILURE in 20m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/54b639369206454c8e68461536acbcb3 : FAILURE in 18m 28s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ad10eb9a93df4466aa752a1da1690144 : FAILURE in 20m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/265e3a7afb614cf28a1d3672d72080a6 : FAILURE in 21m 25s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d92d4c42002b482c8fa4c78e2a2914b5 : FAILURE in 17m 05s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8e97b6442ff74111a306e0a99a019707 : FAILURE in 19m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/339a3868bd8049fa939ffbac79707317 : FAILURE in 18m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0dca24a857e485ba11c2a401345536a : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":18},{"id":"488ae09ec19240446bb5c198f0f6d246e449d24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-06-27 08:59:42.000000000","message":"Patch Set 19: Patch Set 18 was rebased on behalf of \u003cGERRIT_ACCOUNT_34671\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"_revision_number":19},{"id":"c14eea18c5a1a67c26dc113c73b5fa7b199a5e6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 09:10:12.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ca43565ae340758bdbdf83a3412acd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc72fff4ce41433ba7bd55798f093411 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a599967428b54b6f8d1fb3d22225b6df : FAILURE in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"37a79b174c16fd9042a0cebb6151e1c71692fe26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 11:08:33.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c5d1d0663d44cb287e25b56ac6f9afe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b6fe3bb75554409bcbc555ec3f84b0b : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439d8cf633104723a2b9f68a57ebbd52 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f51bb6872b04da89644b276bc925bd2 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74fee882c4dc4b1f902bab1c927ecbe6 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/228c417e78c840308cc7ab4e6e537392 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5cd55fd98f4bdb8dd7cd17768c746b : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c157c26005c14a4dba8b53fd8fd46bf8 : FAILURE in 58m 59s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/11b89294f3c647009cd1e984b5975032 : FAILURE in 1h 52m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b2a1fb1d29f94124b6938c979e24de3e : SUCCESS in 1h 20m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/12e74d885b6b4001934d4cab0372dfc2 : FAILURE in 56m 47s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e2da6d760d5b4b6ab26c46dd39f8c7be : SUCCESS in 2h 06m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/145037d61ef74d1dbb91be13bffa54f8 : SUCCESS in 1h 38m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b8d3d9c6d3cc44768c1f404172816ee6 : SUCCESS in 1h 00m 31s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e4a37be838f54b48b7882758b16c7f6a : FAILURE in 1h 16m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/9fda2e553bc949fdbfd82e4701fced7f : SUCCESS in 1h 42m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c449122a33f419b8e0b590dfbfc1c37 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":19},{"id":"fb546321d5b35fbbf33eab980b3ee0944460c63d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-25 10:00:45.000000000","message":"Patch Set 20: Patch Set 19 was rebased on behalf of \u003cGERRIT_ACCOUNT_34671\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"_revision_number":20},{"id":"bea87884137d32a663905e54ab11971792acb4a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:11:36.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76695dbb15ee498883a81056a3387d2e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d79b26b729604cadb49ce3c7cbaebefe : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea716b841cf14df2ab3663c91b451f36 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"663293f669677aa98bcc8c1a71cf7e67422689f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:52:46.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/814bd9ba78c049d5a157b9d0cd8f5eda\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c36272f4d5a44cc5ad1709ac866707ec : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e2a73082d746fbbe083be26126648f : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44d5c1985b7347d2acc98d025e16a98a : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb7f8b501d86454bbb1815d1640d9357 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47f2d461b0cd4c2ab0793e7461110109 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce987470a38424890c4cc4243067b1b : SUCCESS in 5m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2592410ee5854b79a6deabbe66fc8393 : FAILURE in 42m 01s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/82032a38dd1f475497c76ac42807981a : SUCCESS in 1h 40m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/72a505cb57634ece9e073e981c0f2ad2 : SUCCESS in 1h 20m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0a289764d6494f76b8d69880631c1948 : FAILURE in 57m 18s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5b611d83136f4b6db78bf6569cfefb35 : SUCCESS in 1h 30m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1a5ad1933b3f4f8bb1319f872a1daf45 : SUCCESS in 1h 49m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2b39c7020b274b45b29c7a7ec2109db8 : SUCCESS in 52m 52s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/55171af21f3541d3b695511e8f6e58f5 : SUCCESS in 1h 27m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bdedbd1dd8ee4e11a7ff38f76e4c3e0d : SUCCESS in 1h 10m 23s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3f3dfd954084b8bbf609eaef99e708d : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":20},{"id":"0a884425533825970fe2fdb13d107d215d7f1a75","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-30 14:58:20.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"102cc7d78fa3688b1702395be01a4a5397371dac","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-16 02:08:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"4dc339e8ccc417d6f27b6a339c8a56a24137c12b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-19 04:35:32.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_34671\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"_revision_number":21},{"id":"57be703890723bd1df07c546d9164caa08d82ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 04:48:03.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab3644a0e7f454f84f7dad486d90a11\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80ad57a864264003ab26bcae8e816bd2 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3dd86d1f11ac4e57b442899938a819b3 : SUCCESS in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"38ee8d5651e9a4fe98bba6c6f89e91a069053ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 06:10:30.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b1e8fd133af4db7b157310aae83832b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72a49e0b694342658b2c8724178f9542 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215df51ddd864ed0803726c06f953de1 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b331654c771a4c2d88c9f5f3d164c2b7 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/818d854c39ff460b9f955b103b46ee43 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9d03f7e9c604ac88c6d31aa75f1c8a3 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2151b67ceaa74eb2b3de5285aa125c43 : SUCCESS in 4m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/945427ea300c4e0e95b2261e1856b828 : FAILURE in 40m 30s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/58656f50efa64ffaa559e271bf5d1816 : SUCCESS in 1h 32m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5c42658762b74a6598ad3a1e149a3699 : SUCCESS in 1h 24m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e7f0215830f74504bbb6a2ae4b1cc350 : FAILURE in 38m 00s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e204706c460d49d3b1848e648513bc8f : SUCCESS in 1h 33m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/096793d702c24540a626cd2c805f6bed : SUCCESS in 1h 16m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/54ae3b793071421d8f3111ca68c01773 : SUCCESS in 41m 52s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/91d9eede653748c0822a53789df55674 : SUCCESS in 1h 23m 16s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/34a10a39c9bd4adcb0f53ca38bbc83e8 : SUCCESS in 1h 13m 59s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60ccc42561824ee78004626f08afb06a : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":21},{"id":"f0a7995808370bb679072858f775d31f67d4cc29","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-21 04:17:11.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"0b25763221f4293bc12b77b4127ce54f7b1d0a3e","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-22 02:13:45.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"a28470682130e4494f345426f7ca57fead1e059b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-22 02:16:32.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"70260b0b79f0e9f7431c4ce5a2dd9c791d5458d0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-09-04 05:46:58.000000000","message":"Patch Set 21: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"3212c2cb2bbc9e54bbb0c67838402a06e8eb82cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-09-26 03:22:17.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"54c86386b1361bcf3d09d8268c92c7d93fce2fe6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 03:42:34.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edabdb66226c455b85e814c6052bcd47\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/064a7a15551f4ce0a438f6675a97203f : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/858de171b4834c78bb2032af425f712f : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"26ab6330fd1daa990e9ccfaedd6bd6bc3f74c65e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 05:15:46.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d26a2e9a9bb4e7c8325a690894cb30e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20b2a77074bd430fa8cadf5e90e815a2 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d94d09eed94e35ae4775d201088041 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15e53c64500146e597249c7721f5ccac : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90259d0b315f41a887bcf8e857365b4d : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8310ed752fb41249dc6da34df24ddef : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dcfbb8b66549449ee51df993a2ecc0 : SUCCESS in 5m 28s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f4201fdd6f154924ab9f2b9ae84e051b : SUCCESS in 35m 53s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d9da28c7459c4e5980f31c2bbf332878 : SUCCESS in 1h 51m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e2e41e8be1094c20908fdaff968063c3 : SUCCESS in 1h 29m 38s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4150a41ca30f4b379013e3ba515a7bab : SUCCESS in 45m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ff9c38ce4bc644bc96985443e11a3004 : SUCCESS in 1h 11m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/27979e86f4e5440dbb9977795f76ec91 : SUCCESS in 1h 17m 54s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/9f3c9c5ed2204f4caefd2f5f5393a846 : FAILURE in 41m 56s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/21a943dbda074713b37088fa43271ec5 : SUCCESS in 1h 25m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0183d10644f84cc08bb61775a5371d03 : SUCCESS in 1h 13m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e4ff7294d65742a2889de88197d0ead3 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":22},{"id":"dacb8d39ceb354b25c7c7a86ce4e70916fc3f139","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-09-26 06:10:28.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a894f76e578a553ad60392864876744c59d92ada","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 06:24:46.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c39f57932c24252ba0dc0d2552caeef\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1a0fbfadc0f43ee91f9903f09017bdc : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6ee323480434b5aa30f7c92143fcee1 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"210bbf47dbd043ca891ba10d4422f45ff05d0e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 08:26:01.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b073fde11b7f4b09815ec130865fcdf7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90e564a4ffba4a35b52a50b674356dfa : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29026300e5e045988bcdba9077a5e8fb : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0269c7ea8b824576aa5e9b4d4917c93c : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9569ff2a28a94b1dae5f08b2d8554d36 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bae346a364ea42c6ba6997e3d59ba049 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81ddebfcf4d4b1ca9faa257f4b8625a : SUCCESS in 5m 31s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ebf9023e667345cf9d4008ef50c7f0fd : SUCCESS in 52m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/94580a06cb6d485fa2497ea0c1550452 : SUCCESS in 2h 13m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/659cad2dde744da6918c545a55959c1c : SUCCESS in 1h 33m 57s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8d0a9080b9584720b034724fd8edfaf7 : SUCCESS in 38m 48s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b049a92078dc4636a2432050845d472f : SUCCESS in 1h 39m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/831567a2f84640e38fc014ec52591fa5 : SUCCESS in 1h 25m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/693837b1f5c044eb86810e0799430db4 : SUCCESS in 49m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d5099da8f2ef4a2cba94516f873b75ad : SUCCESS in 1h 43m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/35808e06cd5946009b6e83a5bf6e525b : FAILURE in 41m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/658948c149fe42e8afae96bf18956e57 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":22},{"id":"9c78caa862837a574ee2352af39f51b6241991df","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-09-27 05:49:54.000000000","message":"Patch Set 22: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"9f3d33055a36633632c0c61db1a227805ba64275","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-17 13:03:39.000000000","message":"Patch Set 22: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"52c141e76d6cf19a8635374a4ab4f112db8ff14b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"real_author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-11-05 03:58:08.000000000","message":"Patch Set 23: Patch Set 22 was rebased on behalf of \u003cGERRIT_ACCOUNT_26285\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"_revision_number":23},{"id":"658939e427a3fe682b6465622e36107c1a70899b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 04:11:55.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bec0699fe37403187e745619a497e03\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/990ed9a143f54199a88adf2fbd91aab2 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a94c5c313fb5487fb975a0766cf40cb6 : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"76c484989b1629b270ffbfcb65645b1f34d94fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 05:51:59.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec026aa58dd44b37b93082f4c637b099\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/270ba74c4d6c41c29f24bafd8f4d6acb : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06b4e4289a143aabf8cd170d16b163c : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e55fb8b5421843e5b14d8f8d5c438a70 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63df54761ab84f9e9557e1544d12986b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb9a312d0d74d389a4c31bf3fed0c8f : SUCCESS in 4m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/54aebc906f3141eba1c215d60ce4e8d8 : SUCCESS in 41m 22s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/c82539e38ac64c04951d982e2ab9f51e : SUCCESS in 1h 52m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/963a99287f8e4091a5ee1b72dac41f94 : SUCCESS in 1h 31m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/45ae8743ef714125b4526a6624d1c8e3 : SUCCESS in 38m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/21cd220443634023b9f6d33f83378355 : SUCCESS in 1h 39m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f17ee445539c4845af51dc9423b8ae40 : SUCCESS in 53m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/acf2440c49f94a46b03ae2c09b278240 : FAILURE in 44m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a1757e045ec64db18a7c45e86efe85f3 : FAILURE in 1h 00m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/86d4459b26fd445189d6d306f42ee043 : SUCCESS in 1h 14m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f175b102be94e4aa2973a8ec1cdad34 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":23},{"id":"4518830ed4a56ae282d3d0cb61b2fde505a888ba","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-06 02:26:28.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"9908bdf41f725e06fce5076d4d48ce8e0789287e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 03:00:12.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bd68dfc9a649b9b77691a33c235cbe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e93bcc8ed7684b938bca6388df095582 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56ffc438c1af45949cdde087a04cb3b9 : SUCCESS in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f630548c6038763c680197bcaf65cb50d6753131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 04:57:17.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/436590cdf8614aeb92f9c95fa69cf420\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4d0c5530f32427a8bc9cff7ee2b2874 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6381e9300f244bf9c97e553382a60da : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ddfb6d70444dffad8e2094079e49b5 : SUCCESS in 5m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00d9350431154c6891821ceb42184f31 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfd96257fa64d899b4486d0355daee2 : SUCCESS in 4m 50s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c1847a5ff6ea4d96bcaf107bd8c20c2d : SUCCESS in 1h 00m 10s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ea2fa555b65947be9bf304a948cb7f99 : SUCCESS in 2h 05m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/85b95f607cba45e58c774db614814a45 : SUCCESS in 1h 24m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5d0932d599c44203b42ae48888a63e46 : SUCCESS in 57m 13s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7824609f65fb4f3594e14dc8d8986d0b : SUCCESS in 1h 39m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1d4ac7f0db30453899e5e925033756be : SUCCESS in 1h 33m 25s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/83edd84634c942adb1a8562b4b78db24 : FAILURE in 38m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ba438f83701d41fdba39eb740b6f11bc : SUCCESS in 1h 10m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/8f74467919fa460fbce8c2ae8c7d9dc5 : SUCCESS in 1h 19m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/806d56959270416593c639848d2d1089 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":23},{"id":"3ba174fe4f67ad83bae43776828833e6806e01db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-04 13:38:28.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"d6ccb54b84266df1a13b6cc73ded7b683cbc09d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 13:51:05.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/278eb31381034ae6957001e8e8ded998\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80ccc75e8fe24bb2af8b015464bf69f9 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b474a85e6e9045789661ba834f178344 : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"8b5de4c241ae3e64baa1cf66710255f7ae77987d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:43:05.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/007157621ccd4475ae1ab8e8c2a3760e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/154bfa94c9804f69bc40e013d0f46b85 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e779a40af49e4287a1e0006e1b6c79eb : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68837af656dc419fbb23df53704797e0 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f7e0ff421c54e9884c84b39763c4801 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014b613a23384c08af4a89c3326589d9 : SUCCESS in 5m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e555a8e79947486289a7425b0b8c3dbf : SUCCESS in 56m 12s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/0f63902dcbf445a990b0d0164a179655 : SUCCESS in 52m 58s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d5607b333ccc4650aef59b85381c1f1e : SUCCESS in 2h 03m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/dabc774d218946f3be14538462a41ad5 : SUCCESS in 1h 19m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/67766b8b4d8341fbb7825908dceed85b : SUCCESS in 52m 14s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b39c6b78042d4788b1088d8c45f89ce4 : SUCCESS in 1h 29m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8f1a9081e8c545d2bca86c559c0239b2 : SUCCESS in 1h 17m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a21a61cbcc784fc68f0ca430faf81f32 : FAILURE in 1h 06m 03s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e5c6311d31784341bd38c57a5f03feb6 : SUCCESS in 1h 29m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/405eb5d24ad649308be4faa61e69b833 : SUCCESS in 1h 35m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9db1213fa9aa42c2bae23dffda04f737 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":24},{"id":"5e214e1631331d424d6c568b5fba04d41a7b8f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-11 12:01:53.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"d16f031e4c620059ae9e15179fd09588804de493","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:14:28.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5f62f5d56c94c81a0421998f4693fe2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f603c31dec7846fa85ea448a9f37703d : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c72b581c623146d6abf1b60d20388925 : SUCCESS in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"7fd1c888d2901c35464a233945318219d71d4c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 14:07:30.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0cac72fcc674b6795938a6c96448961\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d23cf001a5404cb7d7ed8d48469f61 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc27b9b5b6d486289750cc9c7d21087 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed61125450443a1b30d93265d39a19b : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93f5b773922241db814258f0a3b081f2 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13afd2ec5f540fba509eaa0c9633d0b : SUCCESS in 4m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/889d5d61b00342c682e650c2849dd7d6 : FAILURE in 33m 29s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a822ee89c1de469faa8deb12e6cd7dcd : SUCCESS in 54m 34s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a389b9e2a4034064a719da6f97b164e2 : FAILURE in 40m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4900d0322eb9467faad6ccba9b372573 : FAILURE in 29m 59s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/95c90490e2b842debd8921756e03d8a7 : SUCCESS in 1h 03m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b66547e40b5a44d9b81ff1003bfcea25 : FAILURE in 29m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5d30a6a5877047ed879eae76f9038f1e : SUCCESS in 2h 02m 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/73a5e25db0fa4a2da6fe9307f25b0b29 : SUCCESS in 1h 01m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d87297ff051d4b8482a814971cbc2b2b : FAILURE in 1h 15m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/28944d3f764f4781b66fd4a0bd57238d : FAILURE in 1h 30m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9546630c1ce49a397d63c7493fa55af : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":25},{"id":"c5e5d0a31c7930c578035228b4a05edbb61dacb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"real_author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-12-17 15:07:33.000000000","message":"Patch Set 26: Patch Set 25 was rebased on behalf of \u003cGERRIT_ACCOUNT_26285\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"_revision_number":26},{"id":"32506d8ba85a694bbdbfbefb99a55f9bcc7bcd16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 15:53:09.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33d72f5454c245b4a3066ccfe7f52f14\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b52fcf732e57467da04c4be19633bc4c : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf7ec2eb90464d82bdd84ba7151be7a6 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f2eea310903b22c19342ad7be16c96548dd6aa5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:31:15.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcfc7cc250254fe38f46d2207c192564\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8699f114f46347a088b9700460da4dc0 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46bd8d2d75b431ca71db6012745a69e : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/468627fe50804d099ae13a8187c5c86d : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6262976402bc4c75a6419d1574eb2e37 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0444587a1f54b1cb8916ab49d627812 : SUCCESS in 4m 58s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c3939dce99234adfa15885d3607d4d2f : SUCCESS in 1h 07m 46s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/7f11422a0e054e04a5338b519e2d5d5b : SUCCESS in 40m 17s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6d5d5702bb3c404eac7f21e95997e60e : FAILURE in 1h 27m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/013f31e597334085944161697dd16cc4 : SUCCESS in 1h 58m 40s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f41b339b7564400fafa3e5bf09f14890 : FAILURE in 1h 12m 55s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/11b54d53296c48e48b471d0edcfb2939 : SUCCESS in 2h 21m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4447a6844f3e4dd69cec813465f523fe : SUCCESS in 1h 47m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1624ae521364431d8e1a98ef1bb1bf34 : SUCCESS in 1h 10m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0552fabec573400e8cddac2ca095eb2d : SUCCESS in 1h 54m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4f9dfcc0b4c2428d867a7b373189bafc : SUCCESS in 1h 13m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe75746c9c3e4b719ce634525616919f : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":26},{"id":"0bc9308f5f42c2c4e8b1f70a55fe9eb755bb377b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-18 08:54:43.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"7a1173148e2bd7c3b16d5a8c65541226e4b03bae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 09:26:30.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3290123358b44247a06ef54cc7d041c1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e34f9c4c944649cc8076a7bd6c266844 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f0e660592f484d81a2c101f1c97ae04c : SUCCESS in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"4feea8389f967822889f7369a61e00727eab5ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 10:52:04.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b715843d35024a809d4ea8268e394945\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d72675502790402c8bb9bf5a1bd028f0 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c431c879224a3d8d13d80406331f59 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fb88707d350480dafa0b786384b3041 : SUCCESS in 5m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13e556b0d54441d49f568847821f08d7 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8729c94f8fc84f429a1334d90f8ae92b : SUCCESS in 4m 50s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/efe5068a6ce5462ba7d23f5b774b85c0 : SUCCESS in 1h 00m 34s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/be87773b0b8b407480694bd62a915ca1 : SUCCESS in 57m 39s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3af9570727d146efbb0914c7b264c858 : SUCCESS in 1h 56m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c070fed26de14b09b53e08792979e843 : SUCCESS in 1h 39m 59s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b34f57390c1b44a49070d152fc5eece0 : SUCCESS in 58m 15s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/43e86041368a439295844bd17fd4c3ef : SUCCESS in 1h 50m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/eb8695b5aed440f7b85f8b8295419271 : SUCCESS in 1h 36m 32s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/cc8aeee2c6344bb1a8195fafedd4f1f8 : SUCCESS in 53m 56s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1195db8d8afb43a8b1b9e46362a6e4d3 : SUCCESS in 1h 27m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/cb81bbd066c74033a21be3d29585a967 : SUCCESS in 1h 23m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d387d723e484e129c0d7604dbf13329 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":27},{"id":"5979f762929c5fc99131ec789c02717e2c70e697","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-24 11:59:41.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"fecc9fb97d7fd6cc8641e1d8d2d0c696d3363728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 12:00:07.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"35c24cc2b2d8905bc9547a4f1dbd527b41b3431f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 13:09:13.000000000","message":"Patch Set 27: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b20f3aa8ca284bc097ba82da0ca50584\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07892e25caaa4500bc10d6d52a586ec4 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7a42eef9d8a48258ed377f84515ed66 : SUCCESS in 8m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c07e1e60dd84a799c64551c549ad939 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4513125a814aa8ae2c0eb86e84145a : SUCCESS in 6m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/a9a3569c533b45d38f3784d3ec1a544e : SUCCESS in 37m 32s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/58982099343f4c74b8a21cc7a08a9d62 : SUCCESS in 34m 58s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/515858703a6d4540bfaa535131565741 : SUCCESS in 37m 11s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5c56876074224b92a9380c252f460fdf : FAILURE in 1h 08m 40s","accounts_in_message":[],"_revision_number":27},{"id":"3ad43b3a89c3141b29a9cc5b1ef5599f5eb2d45c","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-12-25 05:25:26.000000000","message":"Patch Set 27: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"1678ae9155332ca777553fde663bd2682496ffbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 05:40:33.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30c76f4dc2994cf9afb290e4567defa5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71e5a0064dbb4f718fe5b092730508a8 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e998ebeb15f14b3ebe0e6f3dbcfa735f : SUCCESS in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"5ad4fcffe0a62f6527d9532df3565654370b18f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 07:24:32.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a028450058ed46999d4d2961f1e7ce5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4302cfd31993449881f67989052e7a12 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b3c4acce3a4298b6533a953ee9c0f3 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4600f5918954f9ba0f98756e97d342d : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db8ba4958f1144f982b277bcb35c9df3 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817cc7806b824b9ba24cc09a25c51660 : SUCCESS in 5m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/de6da40431fb4714bda04802f02525f3 : SUCCESS in 44m 09s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/6c9ca48f65484055b9f916f73ac2d9f3 : SUCCESS in 41m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4c178e21535e496fbe48020fec1fad47 : SUCCESS in 1h 47m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f45fa133405942d9ad32fff11b3be345 : SUCCESS in 1h 25m 02s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/13c8e227478048a9a60a3d2a7ec0cc98 : SUCCESS in 43m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/be476f59a1f04c8895953fb3fae9b36c : SUCCESS in 1h 37m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2ad6a04ed8184fa6ae6a637419967233 : SUCCESS in 1h 22m 54s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ead9e48baec340cb9231a0f52c3a4539 : SUCCESS in 49m 13s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2c6e0faa6ba84fbebe0ecf62439424ac : SUCCESS in 1h 35m 16s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/c6f5ead2a84548c895f7e72ebed304d0 : SUCCESS in 1h 20m 37s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba6f2fe775b847f29b4957d18b6352aa : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":27},{"id":"a12052910928aecd041cc2fb558975d1bfe24672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 07:25:02.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"e9a36cf5c82bf18cb2033414351c1f7f0324619f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 08:08:28.000000000","message":"Patch Set 27: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79134309d4134f66b79383a01c9ba0f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ee01f1b8c3f4908866771c34ae78549 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85de4f796e0a44a4bfa5bc1c26687b2b : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03c45055fa5c4af4b55991cfaf4dc279 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7418e7416a94c7c919e4aa527bab73d : SUCCESS in 3m 59s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/81ed451235214f0ba071632756cdfc23 : SUCCESS in 32m 33s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/0f03dc39744a49a59ca8c80a100f84c0 : FAILURE in 30m 31s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ce5b4de9790044ef816c1a08152a3b5a : SUCCESS in 42m 05s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d1e6075373b14ec6a031892c6cd9384f : FAILURE in 34m 59s","accounts_in_message":[],"_revision_number":27},{"id":"33e348dd5d83241634c9fd013228e6043aee870e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-25 11:20:56.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"0a82a20ebb3f34d9ce7cf83a722f3a8ada681911","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 11:33:35.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95c06a7248e448a883142635606eeb83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ecfea5b2359449387e22663387e91bf : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/610c3f4048f74d868534b96a8d0ef686 : SUCCESS in 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"0ca0feb8fb9154574429e8158c69243f460dac74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 13:03:23.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35756e8478ee4d74ae3213ee8e949521\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e05056603e41431ea76b2fb2354b8406 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b063081d4184637b9593fff15336a5d : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e30f7fbe78bb4031bccce419e669e16f : SUCCESS in 5m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c6fb06532ae48549274d3a1e06ef04d : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc22701cbfd42fc9d929f1b96ed0577 : SUCCESS in 4m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4a959216a9624dc9b8f9e6debdb9ab36 : SUCCESS in 43m 45s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/5b72f8c0a6e34cea881de4353a9bff82 : SUCCESS in 40m 08s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4f26aa4835464dae99811406ade53b26 : SUCCESS in 1h 41m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fdc655458c0541f0b91103c5c5ee3c28 : SUCCESS in 58m 10s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/afcd8958d07b48ec9ad1b5e1a1122c39 : SUCCESS in 41m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/33e43e7ae4be48b18f1ff17b95849bdf : SUCCESS in 1h 32m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9ea3e95c21dd47e29b8741ce3487daec : SUCCESS in 1h 17m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/378ee13f6394447999d912f24a5a22fd : SUCCESS in 44m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/75b36debe8954a46a3d7b874247843f2 : SUCCESS in 1h 33m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/124c895ec6974c51b4357d3a08b3f919 : SUCCESS in 1h 14m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/52dd9e9e80e3468facd311c46ec3bbe5 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":27},{"id":"173cdc72514d461e825377b19de3aaab1818c001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 13:03:52.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"76c0f28e80d62043d981ae15e47e48f0b2b13f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 13:52:41.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/769a1f368fb649e2b7cae28c8be906ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f15049cec07479ab1de94649e3d8f22 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f9b0023f56b47909144f5fa1c3cc27d : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c8a1a7b0cfb4eb698ecf463cdcf4009 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221ceba4527a4ff6b54016d61d692b5c : SUCCESS in 4m 41s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/162c67c800c842a98a6331c0b37654c1 : SUCCESS in 46m 29s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/278485ee64dc40ad890347239081199d : SUCCESS in 43m 06s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f45b50e9d0d64945b7b49733d7e6a527 : SUCCESS in 45m 08s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/df607435fa094e0db66d047e5e760d88 : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":27},{"id":"6d60b5ec3ea74d70a635b9349b18437d2515345d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 13:52:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":27},{"id":"b008dd97fca0bb9dffd75a708dc1efb791d30f8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 13:53:40.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4449501faaee40f3b2ff98c983734e3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a788ab873f1451a825ac2bf2510bd06 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"89274df2aff829cf73920e030679c6fb0bd461e3","revisions":{"081b1a85d32bd989f66ec4a71544001587dc38aa":{"kind":"REWORK","_number":1,"created":"2023-03-10 06:53:06.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/1"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/081b1a85d32bd989f66ec4a71544001587dc38aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/081b1a85d32bd989f66ec4a71544001587dc38aa"}]},"branch":"refs/heads/master"},"37a387a792d15611c18c75d884f1cc847068f910":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-10 06:53:47.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/2"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:53:47.000000000","tz":0},"subject":"Working on missing postgresql features.","message":"Working on missing postgresql features.\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37a387a792d15611c18c75d884f1cc847068f910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37a387a792d15611c18c75d884f1cc847068f910"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0e97f25c6b11f1edbbdf9deef5464bc4e180013e":{"kind":"REWORK","_number":3,"created":"2023-03-20 11:06:10.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/3"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-20 11:03:32.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0e97f25c6b11f1edbbdf9deef5464bc4e180013e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0e97f25c6b11f1edbbdf9deef5464bc4e180013e"}]},"branch":"refs/heads/master"},"23e9fdd2c0972faa0125896287faba8139a838b7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-03-20 11:06:59.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/4"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-20 11:06:59.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/23e9fdd2c0972faa0125896287faba8139a838b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/23e9fdd2c0972faa0125896287faba8139a838b7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b3b5aef3d24816665f6161aac554f40798675069":{"kind":"REWORK","_number":5,"created":"2023-03-20 12:56:32.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/5"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-20 12:52:17.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3b5aef3d24816665f6161aac554f40798675069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3b5aef3d24816665f6161aac554f40798675069"}]},"branch":"refs/heads/master"},"4cfa9b15461c38f8b593ba296c5c829ad1abcf83":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-03-20 12:57:31.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/6"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-20 12:57:31.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cfa9b15461c38f8b593ba296c5c829ad1abcf83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cfa9b15461c38f8b593ba296c5c829ad1abcf83"}]},"branch":"refs/heads/master","description":"Edit commit message"},"757e6cd2cdd28cec5e39dbf755abf4f0c819f1a4":{"kind":"REWORK","_number":7,"created":"2023-03-21 13:20:15.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/7"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-21 13:17:00.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/757e6cd2cdd28cec5e39dbf755abf4f0c819f1a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/757e6cd2cdd28cec5e39dbf755abf4f0c819f1a4"}]},"branch":"refs/heads/master"},"6483ea98b5549f7c66d05b8f1046e0ae50207d18":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-03-21 13:20:58.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/8"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-21 13:20:58.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6483ea98b5549f7c66d05b8f1046e0ae50207d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6483ea98b5549f7c66d05b8f1046e0ae50207d18"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6c50f4e907af41d141296aa9298f32c04369bcd5":{"kind":"REWORK","_number":9,"created":"2023-03-22 06:46:13.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/9"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-22 06:43:43.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c50f4e907af41d141296aa9298f32c04369bcd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c50f4e907af41d141296aa9298f32c04369bcd5"}]},"branch":"refs/heads/master"},"510bbefb6f66bc9f9e341993f742d23dc82095d6":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-03-22 06:46:42.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/10"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-22 06:46:42.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/510bbefb6f66bc9f9e341993f742d23dc82095d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/510bbefb6f66bc9f9e341993f742d23dc82095d6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9b2f75486a702b999ced75ac5eb71e3b665049b4":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-07-20 03:18:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/71/877071/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/11"}}},"commit":{"parents":[{"commit":"023d15b302b51d6103f5d95248762599ee776007","subject":"Merge \"[CI] fix devstack install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/023d15b302b51d6103f5d95248762599ee776007"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-07-20 03:18:03.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9b2f75486a702b999ced75ac5eb71e3b665049b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9b2f75486a702b999ced75ac5eb71e3b665049b4"}]},"branch":"refs/heads/master","description":"Rebase"},"440168e175a46829c5dc05686ff51d65a9d4e4e5":{"kind":"REWORK","_number":12,"created":"2023-07-20 06:57:28.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/12"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-07-20 06:53:45.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/440168e175a46829c5dc05686ff51d65a9d4e4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/440168e175a46829c5dc05686ff51d65a9d4e4e5"}]},"branch":"refs/heads/master"},"8ad0271e51e44b5aa5fc7edc2160b46ca1b1f4cf":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2023-07-20 08:04:59.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/13"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-07-20 08:04:59.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8ad0271e51e44b5aa5fc7edc2160b46ca1b1f4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8ad0271e51e44b5aa5fc7edc2160b46ca1b1f4cf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"82742868b5ba05d119905c72607232cf78224373":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-04-02 03:18:11.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/71/877071/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/14"}}},"commit":{"parents":[{"commit":"a2775ec2aa459a0275ac86180ce0f091ca8ad384","subject":"Merge \"Get rid of nose and Proboscis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2775ec2aa459a0275ac86180ce0f091ca8ad384"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2024-04-02 03:18:11.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/82742868b5ba05d119905c72607232cf78224373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/82742868b5ba05d119905c72607232cf78224373"}]},"branch":"refs/heads/master","description":"Rebase"},"3647b281e037a436bd43b6f0c9922cd33dc37591":{"kind":"REWORK","_number":15,"created":"2024-04-02 07:01:32.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/15"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-04-02 07:01:14.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3647b281e037a436bd43b6f0c9922cd33dc37591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3647b281e037a436bd43b6f0c9922cd33dc37591"}]},"branch":"refs/heads/master"},"ca0f92ecdcac953c60192b954d5ee1d1989cd5a1":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2024-04-02 10:59:00.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/16"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2024-04-02 10:59:00.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca0f92ecdcac953c60192b954d5ee1d1989cd5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca0f92ecdcac953c60192b954d5ee1d1989cd5a1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bfd0bc325ba4bf1ab1c0fa4264eb1cc5f7a65326":{"kind":"REWORK","_number":17,"created":"2024-05-17 08:13:08.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/17"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-05-17 08:12:54.000000000","tz":0},"subject":"Working on missing postgresql features * Db list operation added for postgresql * User list operation added for postgresql * User create operation added for postgresql * User delete operation added for postgresql * Db create operation added for postgresql * Db delete operation added for postgresql * User grant access operation added for postgresql * User revoke access and update attributes operations added for postgresql * Common/db/models collate and character_set setter functions are updated.","message":"Working on missing postgresql features\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bfd0bc325ba4bf1ab1c0fa4264eb1cc5f7a65326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bfd0bc325ba4bf1ab1c0fa4264eb1cc5f7a65326"}]},"branch":"refs/heads/master"},"cd2ea4c104aa438442ba130b69e78dab196cd85a":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2024-05-17 08:13:28.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/71/877071/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/18"}}},"commit":{"parents":[{"commit":"61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a","subject":"Merge \"Fix tox cover check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f5bb8c2fee8bf26cdd531b0c5bccaf910f281a"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2024-05-17 08:13:28.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd2ea4c104aa438442ba130b69e78dab196cd85a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd2ea4c104aa438442ba130b69e78dab196cd85a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7c8f8bef3532921017d5da06a242793bcfd70c3a":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-06-27 08:59:42.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_uploader":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"ref":"refs/changes/71/877071/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/19"}}},"commit":{"parents":[{"commit":"114601edba9fc736af43a3f2705552d8c746115c","subject":"Introduces Alembic database migration tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/114601edba9fc736af43a3f2705552d8c746115c"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2024-06-27 08:59:42.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c8f8bef3532921017d5da06a242793bcfd70c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c8f8bef3532921017d5da06a242793bcfd70c3a"}]},"branch":"refs/heads/master","description":"Rebase"},"e4901c467d37cb135837a52484921601a80d44d2":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-07-25 10:00:45.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/71/877071/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/20"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2024-07-25 10:00:45.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4901c467d37cb135837a52484921601a80d44d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4901c467d37cb135837a52484921601a80d44d2"}]},"branch":"refs/heads/master","description":"Rebase"},"bd8e22e281d50e245c3c35aa8ebd04e04789664b":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-08-19 04:35:32.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"real_uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/71/877071/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/21"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2024-08-19 04:35:32.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd8e22e281d50e245c3c35aa8ebd04e04789664b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd8e22e281d50e245c3c35aa8ebd04e04789664b"}]},"branch":"refs/heads/master","description":"Rebase"},"08614ea19d2d895f2fde0b2e90da97a0e441d07f":{"kind":"REWORK","_number":22,"created":"2024-09-26 03:22:17.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/71/877071/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/22"}}},"commit":{"parents":[{"commit":"af2bea2dcc03f91cfb6fc8e8a8217be6fad501da","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2bea2dcc03f91cfb6fc8e8a8217be6fad501da"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-09-26 03:20:38.000000000","tz":480},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/930276\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08614ea19d2d895f2fde0b2e90da97a0e441d07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08614ea19d2d895f2fde0b2e90da97a0e441d07f"}]},"branch":"refs/heads/master"},"d33da69510c72d2fb486c668e19abebae7ec3aa5":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2024-11-05 03:58:08.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"real_uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/71/877071/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/23"}}},"commit":{"parents":[{"commit":"3ff639047eea34c416872014639258d9ecdf067b","subject":"Remove Python 3.8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff639047eea34c416872014639258d9ecdf067b"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-11-05 03:58:08.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/930276\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d33da69510c72d2fb486c668e19abebae7ec3aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d33da69510c72d2fb486c668e19abebae7ec3aa5"}]},"branch":"refs/heads/master","description":"Rebase"},"85982204a78ec1c6d2111855409dc73260109400":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-12-04 13:38:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/71/877071/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/24"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-04 13:38:28.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/930276\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/85982204a78ec1c6d2111855409dc73260109400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/85982204a78ec1c6d2111855409dc73260109400"}]},"branch":"refs/heads/master","description":"Rebase"},"828235b9830697e51be656cef77128e360859d0d":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2024-12-11 12:01:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/71/877071/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/25"}}},"commit":{"parents":[{"commit":"17b7c60270297b862e363449b03bf6ce8a9c4693","subject":"Merge \"CI: push images to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/17b7c60270297b862e363449b03bf6ce8a9c4693"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-11 12:01:53.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/930276\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/828235b9830697e51be656cef77128e360859d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/828235b9830697e51be656cef77128e360859d0d"}]},"branch":"refs/heads/master","description":"Rebase"},"51347967ed88df937c3cfd755407b1272202f797":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2024-12-17 15:07:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"real_uploader":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"ref":"refs/changes/71/877071/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/26"}}},"commit":{"parents":[{"commit":"ff502ae696579ebe14a223fb309a5acdc963ebd7","subject":"[CI] Follow up: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff502ae696579ebe14a223fb309a5acdc963ebd7"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-17 15:07:33.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/930276\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/51347967ed88df937c3cfd755407b1272202f797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/51347967ed88df937c3cfd755407b1272202f797"}]},"branch":"refs/heads/master","description":"Rebase"},"89274df2aff829cf73920e030679c6fb0bd461e3":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2024-12-18 08:54:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/71/877071/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/877071/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/877071/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/877071/27"}}},"commit":{"parents":[{"commit":"0138ce30e256a9f3a8fc3405a16051aa96294d57","subject":"Using quay.io registry in project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0138ce30e256a9f3a8fc3405a16051aa96294d57"}]}],"author":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-03-10 06:47:14.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-18 08:54:43.000000000","tz":0},"subject":"Working on missing postgresql features","message":"Working on missing postgresql features\n\n* Db list operation added for postgresql\n* User list operation added for postgresql\n* User create operation added for postgresql\n* User delete operation added for postgresql\n* Db create operation added for postgresql\n* Db delete operation added for postgresql\n* User grant access operation added for postgresql\n* User revoke access and update attributes operations added for postgresql\n* Common/db/models collate and character_set setter functions are updated.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/930276\n\nChange-Id: I4c000363dd046fc82f099de0a30c16b1f01d6d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/89274df2aff829cf73920e030679c6fb0bd461e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/89274df2aff829cf73920e030679c6fb0bd461e3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
