)]}'
{"id":"openstack%2Ftrove~878344","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b","project":"openstack/trove","branch":"stable/2023.1","topic":"story/2010467-stable/2023.1","hashtags":[],"change_id":"I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b","subject":"Fixes permission problem when restoring backup","status":"MERGED","created":"2023-03-23 13:36:52.000000000","updated":"2023-03-31 07:39:12.000000000","submitted":"2023-03-31 07:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878344-story/2010467-stable/2023.1","cherry_pick_of_change":871951,"cherry_pick_of_patch_set":2,"meta_rev_id":"4fbefc2815467288b7cce9c461fcd97ad26cf644","_number":878344,"virtual_id_number":878344,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 07:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-31 05:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-31 05:52:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 13:36:52.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"CC"},{"updated":"2023-03-23 13:44:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-23 15:41:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac29f9af0c25d381e4162ea0950b3eb200e3b27a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-23 13:36:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"68bf5fab5bb3d1c46ddbfc2d9d8f6b7f6cc0253e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 13:44:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d49a7fb4a384345aacbe78699e4e0b1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6f44e8f8bad4c54a81b788fa878c0fc : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0300bd76639a493ab2107cf05c586dbf : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70e07011b4c81da1521ae2f07eb05627d573c42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:41:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0ef5d752b14f738468d29bb994dc26\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc5e32ee8a544e528c6b92f5a2ce1a87 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63406807a4ab45dbba784c10769cf291 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85f5338ce12747cda82a5b9b84ff3d92 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c738cc1471c403aaf3362a49d456611 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef20d690b7f417ead9f48a53abdeb73 : SUCCESS in 5m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b95351f1907944ae9d96d7e7e566f69d : SUCCESS in 5m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d9a7fd81d1d84e67845022536ef2580a : FAILURE in 2h 02m 12s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/3cd9b0d7a12a48db9ead85de0385efd2 : SUCCESS in 1h 52m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cf310bac69d439da6d5fba7c9460360 : FAILURE in 1h 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb52c596fd1c27aea6f46f7c1fc9ca75458b7ea6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-31 05:52:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50b8c6a0838e711247885fffad9cdcf0b4ffb4d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 05:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14c1a9aa2df042ba432c35246cf902351f3515af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:38:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d016d91f898e4b388d902fbf50a6803c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43976cd91bd445a7beea123b450ac626 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa59f7f214c4660bed015d473e9778f : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68e5d2fa8eb847dfbac05c8e48e795b8 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d16b3eee0e4940bb9ab83a8185cbae : SUCCESS in 5m 28s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e2ca57b62b924d18a7a8637e3a6a296e : FAILURE in 1h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cff6cda503883cdc453d4eba79303483f414e32b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4fbefc2815467288b7cce9c461fcd97ad26cf644","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:39:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c40504d0b14ca291579fc54ffacac3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460bb35640f84996a6275da96ed4cdc8 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edbdbe367d5eebdbde82e5905f365e9c693ed364","revisions":{"edbdbe367d5eebdbde82e5905f365e9c693ed364":{"kind":"REWORK","_number":1,"created":"2023-03-23 13:36:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/44/878344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/44/878344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/878344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/878344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/878344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/44/878344/1"}}},"commit":{"parents":[{"commit":"8d54913b36fc3d8a80fb02f78dc541f0f08f763b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d54913b36fc3d8a80fb02f78dc541f0f08f763b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-27 05:34:49.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-03-23 13:36:52.000000000","tz":0},"subject":"Fixes permission problem when restoring backup","message":"Fixes permission problem when restoring backup\n\nThis PR changes the owner of the file used by the docker container that\nresets the mysql root password.\n\nOriginal problem is the failure to restore mysql:5.7.x backup. The error\noccurs because the container process tries to update the file that owned\nby the others.\n\nStory: 2010467\nTask: 47017\nChange-Id: I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b\n(cherry picked from commit 553c578923d30d0eb15ad5dbd0a2c583cae274a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edbdbe367d5eebdbde82e5905f365e9c693ed364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edbdbe367d5eebdbde82e5905f365e9c693ed364"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
