)]}'
{"id":"openstack%2Ftrove~878345","triplet_id":"openstack%2Ftrove~stable%2Fzed~I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b","project":"openstack/trove","branch":"stable/zed","topic":"story/2010467-stable/zed","hashtags":[],"change_id":"I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b","subject":"Fixes permission problem when restoring backup","status":"MERGED","created":"2023-03-23 13:37:54.000000000","updated":"2023-03-31 07:39:19.000000000","submitted":"2023-03-31 07:38:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878345-story/2010467-stable/zed","cherry_pick_of_change":871951,"cherry_pick_of_patch_set":2,"meta_rev_id":"f24e917141d724a202978d9ed2c799d8c126face","_number":878345,"virtual_id_number":878345,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 07:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-31 05:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-31 05:52:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 13:37:54.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"CC"},{"updated":"2023-03-23 15:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d4466f1b40c6e2d6ae010a35dad28cac347034c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-23 13:37:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"274d0c53d91faf2ff1a943a4d0180805bb618ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8c26dd9999441e8a8a073251d96401\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/519a38415fa64bc5bf16dc7d7ef9581a : FAILURE in 6m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1638344d3343d9b431d57e39c703a7 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220f0fa0db53475da7b30cfa38535629 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/befc71b23f7c456cb780362b7ecc346d : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3886cba79f7543c89f9441b08a846f97 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a09882ffca4103b14b887d1bbc1192 : SUCCESS in 5m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02a2b934960f41a081f7d0796d127c40 : SUCCESS in 5m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0dc049e9d8bb4a7eaffc84a26c27da43 : FAILURE in 34m 26s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c05970ac82184d9fad7e59bf87da7c61 : SUCCESS in 1h 19m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af3abd5fb3fe4045a00d9db92bde1f3c : FAILURE in 44m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9d043e46a4c54456b2d7854f33b1a21d : FAILURE in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42a627d8afb6eedc9bd828cb308940b9e9b88183","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-31 05:52:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4f626496ce6ac84a79e00ee7ddd865781528a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 05:53:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5519db180997bf2ad815527b2df22c6c74594847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:38:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a30fc5081a934dd4a9a7972eebd1f6de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25baf10f5cbb4d9bad23a3a2d99dea0c : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6493cc2567bc4416b5c45752dbfb7e48 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b10bb5442b54201af2d56b3ad0d1bc1 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee511df09db1429e83b74f8f4b3596f4 : SUCCESS in 6m 51s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0d117e568d394271852a50dec086fc4d : FAILURE in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df4e39a31d74d318e19d075d00c64e2e785eec10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:38:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f24e917141d724a202978d9ed2c799d8c126face","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:39:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e144ed21514645bee243567db44136\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac53217dfa3d411da15a2558dc00c585 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c8b90531402e8cd6ae879ab636fd636c123ebe5","revisions":{"7c8b90531402e8cd6ae879ab636fd636c123ebe5":{"kind":"REWORK","_number":1,"created":"2023-03-23 13:37:54.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/45/878345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/45/878345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/45/878345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/45/878345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/45/878345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/45/878345/1"}}},"commit":{"parents":[{"commit":"d99811a31f0130039046dca5fc60d379f2e56476","subject":"Change settings for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d99811a31f0130039046dca5fc60d379f2e56476"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2023-01-27 05:34:49.000000000","tz":540},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-03-23 13:37:54.000000000","tz":0},"subject":"Fixes permission problem when restoring backup","message":"Fixes permission problem when restoring backup\n\nThis PR changes the owner of the file used by the docker container that\nresets the mysql root password.\n\nOriginal problem is the failure to restore mysql:5.7.x backup. The error\noccurs because the container process tries to update the file that owned\nby the others.\n\nStory: 2010467\nTask: 47017\nChange-Id: I3d8ebaac4ee5e13af83c305ce28b51e4442d8c8b\n(cherry picked from commit 553c578923d30d0eb15ad5dbd0a2c583cae274a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c8b90531402e8cd6ae879ab636fd636c123ebe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c8b90531402e8cd6ae879ab636fd636c123ebe5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
