)]}'
{"id":"openstack%2Ftrove~878858","triplet_id":"openstack%2Ftrove~master~I8845918929aeb3c35270be9148120a1f445e3fa9","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I8845918929aeb3c35270be9148120a1f445e3fa9","subject":"Using 5.7.29 tag for mysql image.","status":"MERGED","created":"2023-03-29 12:53:50.000000000","updated":"2023-03-30 02:54:40.000000000","submitted":"2023-03-30 02:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878858","meta_rev_id":"684c8e25bb8af006249378ff5fc35d3ccfd6d0f0","_number":878858,"virtual_id_number":878858,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-30 02:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-30 01:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-30 01:26:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 13:07:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 14:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6d9a72a30d06d4ba56ef99d001e21403d9c6edb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-29 12:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f265d90722950adb1f2db645261ad43ffa6bad3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 13:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30242e31825f4c93afe21604ed21dd57\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0623751bc7b4acd8c38a345d770507b : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02e9c1ca88a24cc1a4c5986eeed71d62 : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6eea3d4e4bf1730dbbc9814aed66f4518b39a802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 14:28:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/618957c6a6064f5b8420b21d8de3a380\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f13a32e118df4332b59c76ef6c77959c : FAILURE in 6m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeff8fab2d4f408497d4adad28ed960f : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca865338375b4385baee779684812451 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6228ae67a3847bc97b24f0cd700ff41 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9c36860dc848a1b8d8d8062116f767 : SUCCESS in 6m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10987f90dce24ca9936339b48d7de9ea : SUCCESS in 5m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f0778666319149f7b0001551f823ea93 : SUCCESS in 1h 22m 22s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0a9e758de1f840df9aa11d93b4c48ed1 : SUCCESS in 1h 15m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d3165157bdd49c48bc4733234aee4c7 : FAILURE in 1h 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfcf017f843f266bc610a935a2ef62eed5272ffc","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-30 01:26:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b890a2de0985cccb64dcb8cd278e4aba2ade4eac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 01:26:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f84adace98c7f8408877356232d2a53b66a16006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 02:53:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ff37b813d5947b2884e60aefec42249\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d708721fbf9b4a6cb3316e639525afbd : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e234e3f0c8f94abf80da7d73ce810b64 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/775d1c84744b460b92c70b56f1be001c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab153d5b2a7d4fb0843278a2ffd2c342 : SUCCESS in 4m 33s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e8658a26b1c64a3ebf282f879b724cc1 : SUCCESS in 1h 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af932261f6ff08398f8fe95581a4acae925fbb48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 02:53:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"684c8e25bb8af006249378ff5fc35d3ccfd6d0f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 02:54:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f249bd5b8745acb72e7efd0f5c4dd8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdfd0cba559c4dc89bb21ad92a1f1355 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db24fc01db3b010e763352593f91ea17af71bfd0","revisions":{"db24fc01db3b010e763352593f91ea17af71bfd0":{"kind":"REWORK","_number":1,"created":"2023-03-29 12:53:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/58/878858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/58/878858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/878858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/878858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/878858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/58/878858/1"}}},"commit":{"parents":[{"commit":"ab68c2d6803d996c561af70b30f22e780f79781f","subject":"Merge \"Ignore more files for zuul jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab68c2d6803d996c561af70b30f22e780f79781f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-03-29 12:34:43.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-03-29 12:34:43.000000000","tz":480},"subject":"Using 5.7.29 tag for mysql image.","message":"Using 5.7.29 tag for mysql image.\n\nmysql:5.7 image always links to the lastest mysql 5.7 image,\nthis may not be stable for our CI test.\n\nChange-Id: I8845918929aeb3c35270be9148120a1f445e3fa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db24fc01db3b010e763352593f91ea17af71bfd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db24fc01db3b010e763352593f91ea17af71bfd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
