)]}'
{"id":"openstack%2Ftrove~878859","triplet_id":"openstack%2Ftrove~master~Icb0fe4b7357805410de9e8c9f6c9c8a5284827c0","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-04-29 02:24:11.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Icb0fe4b7357805410de9e8c9f6c9c8a5284827c0","subject":"Test 5.7.29 for mysql datastore","status":"MERGED","created":"2023-03-29 12:57:47.000000000","updated":"2023-04-29 04:09:54.000000000","submitted":"2023-04-29 04:07:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878859","meta_rev_id":"ff0131e63588d9ce5507a77ea2ae7e352573ca68","_number":878859,"virtual_id_number":878859,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-29 04:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-04-18 14:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-29 02:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-29 02:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 13:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 13:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 14:30:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"d849580c17ddbeb884e54fbb0acb45f4dc3538cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-29 12:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"076d5722b50faab593d6616d830dc2615840a258","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 13:30:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5217d13ef34b4f3abb0f974a3c0cf02f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3f058692d75402da3b178e2b1c96c06 : SUCCESS in 29m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ba1659dd9c34aa98c2251eb02af23d9 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"731927e1f0839b6a88c332ce3db66504ae3e319b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 13:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a3634c7e6d405181a0b952922dad0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99f9b3afa6154476b790e156b110d100 : FAILURE in 7m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f3ee2270e14d72b7082ce3cbf59872 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db7cfb3b73f64ffa870260de3885172a : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2b02614c5f94da4bcd5d508077fc61e : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d21f37ca0a043a1888c329be5a0f7f6 : SUCCESS in 5m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6a813181fde4b6ca8f1a6a46293edf3 : SUCCESS in 5m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/261ba830a29a41d48428920766ff21cf : FAILURE in 38m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83b0269c7aeea325540185f6ed62f90601d98b8a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-30 07:43:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fcafa82316d06101bcc4e45246e541817c5f475","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 08:12:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4df9d59e1ac40eba12de8709d0824c8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32aa74cd46d04848a5dc7b2827bf2873 : SUCCESS in 27m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/933ecb823221498d9c79bffdd774a283 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd674b5e275889eee65beadfffc218dc6ed87f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 08:20:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5326b524d3ba49949e94a1466088b295\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52b48a84ca204a12972c5150f9416153 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa42aab540634161a308bb70635abf7b : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3d7d6b05a194697946af99b92450fbf : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf2c984973a54083a35993104f39b3d5 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfb356445524dacb3e515cd62f411d4 : SUCCESS in 5m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a616669b698b4e09986e8498c656fcac : SUCCESS in 5m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4ade2aede7694fee8ea574e920a77630 : FAILURE in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e79c7b4e3479d17b21df1f110845ab316feb10c1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-31 03:27:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c44a45b344b5bedc19e3947e333576e055cf0736","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 03:44:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a702345fd438426ca9a388c20de30d37\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/422400be16f5459bba380a5ea8e55094 : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d00d9f892f954eabbdff2ec4dae4f6ef : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0480e5868c714f72fbff4b777ff8c05ec4fba02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 04:08:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be220b6bee2a4691a225f9a9d9df44e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7dff34fa8604cfd9fca116a6fc2241b : FAILURE in 7m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f53a5fe75dc4efca41e1e2a5b32bf39 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29514f20d77343a0b22089c0bba0fcc7 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a767b53feeb7450fbeecd1417a5a8327 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63be4379a0bc4f42a86dcd82c290114e : SUCCESS in 5m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c4bee894d6a4cc5828b6f0c7f1bd10b : SUCCESS in 4m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3955000b7c0542e2afce83366ab9b29e : FAILURE in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cccee52d70c14b8ada574c674085e61eb7df9eb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-03-31 05:51:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5068ab6e7698bac86ce7789903343909c02b4ded","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 06:13:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9562a568888b452aa10550dff0d82bb6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f47fd4e97984aeeb96ef9d8ffa854e0 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e379c4371c1641d6b56c54298248ef57 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04def502e0c71050b79fee5566a6036ee0489a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 06:43:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e230fb37d0f94508954668eecabb8963\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9259e764af544c9888e3feedf71eed8 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75da14599f3f4c119a8284764d71297e : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d606b5f9a64e9487ef289060d13292 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b21ae0cb79b4b86876da315c6637047 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642f30ba26ce4c47867c99cf35c085c0 : SUCCESS in 5m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10e44d809919442589e53f110289c77e : SUCCESS in 4m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b070132d79db438ab41ec95155c35ea7 : FAILURE in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44c1e541b99dee46ab7baa6fa94aa9838afa9f8f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-04-18 09:59:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed30aec77338b477e68deb51df2dfabce51189ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 10:06:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a3b00ed2b8a4cbd96a1c1148305d648\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2b7c6b65c644af2b84f261b6b8624b2 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93e80240586c4ec8bee5667edb7c9682 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d70bb47920d6cefac4d6c93fef17feb32e7ab2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 11:50:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e85dca5478974dc6a5812b157ff30a3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af57701bd1b0400fbc14014ba03bedb2 : FAILURE in 10m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb5671e7fd74baa88496a22c156ed2d : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1a9dc96eca3442c999a10aa01a873f0 : SUCCESS in 8m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2f13569d2db4748a17e20af6c66628f : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8b9411bd39477896a7137f10bad8b4 : SUCCESS in 5m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d6315abb8c64588a40e41865abb15aa : SUCCESS in 5m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/03282440174a45709f3d9df68523d72a : SUCCESS in 1h 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30506b11abe1465c56357b66b7896ae25005e42c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-18 14:30:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bf762931e2bb46c0f1026391c1703554e035d892","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-04-29 02:24:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f682f78066d394c7f873f0607723f48d86dcc3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 02:25:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9719d2f9037e95ab9b8835d3c17b8e71f1329038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 04:07:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/533f34af81a54d258bd3eea902678dd7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90732dd787c54cc79a8190abd13789c5 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bac5535e5de419bb6c22badb62aecc3 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37956f5bf2b7431189ac79a30dae34f0 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf0103ae85c4202bbab004cfdb9fc6d : SUCCESS in 5m 29s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6e182f3d9a2c4e08921adc80f24f52e8 : FAILURE in 1h 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b545d2d161d670c2ff9666964aa2220dca58b976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 04:07:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff0131e63588d9ce5507a77ea2ae7e352573ca68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 04:09:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51bdd463c9db485e8d24449ef4ee39a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3410b3caa3aa431a8e4f49c282cd65cb : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d4b17339e2fe4ded2672abf7fc055d34b21680e","revisions":{"6b73b907b69cdc0d5f317e1448d0143f5ce00b9a":{"kind":"REWORK","_number":1,"created":"2023-03-29 12:57:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/59/878859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/878859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/878859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/878859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/878859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/878859/1"}}},"commit":{"parents":[{"commit":"db24fc01db3b010e763352593f91ea17af71bfd0","subject":"Using 5.7.29 tag for mysql image.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db24fc01db3b010e763352593f91ea17af71bfd0"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-03-29 12:55:33.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-03-29 12:55:33.000000000","tz":480},"subject":"Test 5.7.29 for mysql datastore","message":"Test 5.7.29 for mysql datastore\n\nwe test a exact mysql version in CI\n\nChange-Id: Icb0fe4b7357805410de9e8c9f6c9c8a5284827c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b73b907b69cdc0d5f317e1448d0143f5ce00b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b73b907b69cdc0d5f317e1448d0143f5ce00b9a"}]},"branch":"refs/heads/master"},"7d4b17339e2fe4ded2672abf7fc055d34b21680e":{"kind":"REWORK","_number":2,"created":"2023-03-31 05:51:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/59/878859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/878859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/878859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/878859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/878859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/878859/2"}}},"commit":{"parents":[{"commit":"db24fc01db3b010e763352593f91ea17af71bfd0","subject":"Using 5.7.29 tag for mysql image.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db24fc01db3b010e763352593f91ea17af71bfd0"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-03-29 12:55:33.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-03-31 05:51:26.000000000","tz":480},"subject":"Test 5.7.29 for mysql datastore","message":"Test 5.7.29 for mysql datastore\n\nwe test a exact mysql version in CI\n\nChange-Id: Icb0fe4b7357805410de9e8c9f6c9c8a5284827c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d4b17339e2fe4ded2672abf7fc055d34b21680e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d4b17339e2fe4ded2672abf7fc055d34b21680e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
