)]}'
{"id":"openstack%2Ftrove~879738","triplet_id":"openstack%2Ftrove~master~I4dcd2f4b556d3ea95b3562580d5a18899fb1100b","project":"openstack/trove","branch":"master","topic":"psql-fix","attention_set":{},"removed_from_attention_set":{"35169":{"account":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"last_update":"2024-04-02 03:15:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4dcd2f4b556d3ea95b3562580d5a18899fb1100b","subject":"Fix psql connection","status":"ABANDONED","created":"2023-04-06 12:31:06.000000000","updated":"2024-04-02 03:15:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1426613aa4c51af1e9aa5295b59f7c48e229fa66","_number":879738,"virtual_id_number":879738,"owner":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-06 14:27:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-16 03:46:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-06 12:37:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-06 14:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 03:46:43.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"208fdf1dc50e79bb44e7fd74253c1d10122b234b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"date":"2023-04-06 12:31:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06ebf12c0798108f003263d09fd0fa99e01f3592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 12:37:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4d8471b4bb64295b5ef36a1240ee41d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6cdf0b419b54221a0299b19a44ff905 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98fe6a987dfd42339d5c8bb78af32622 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67637b9a337838ddc82000b4d472c54168d4664c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"date":"2023-04-06 12:44:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f3ba487a752e955dc2430486716e000dca33d91b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 12:51:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb2cf3956c6341f7b0c8f6530d4ae195\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e8cd62bdeb94221996dda4e652cd544 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5deb85eaed94315a30537a2a254bd07 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eee4d60947cdd5d6b3de69b7332ff8e5a8f5e026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:27:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/109ae1cb50d74891890f50af991c49af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1230275670d047c580925e2867addaf7 : FAILURE in 6m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3f0892438a4a36859b2670dc2b9328 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37d04038c3804c76b4ac06ca549e9307 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b39432fe19584ccb8071e582085f7a32 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e364d908db43b0be5a7cb677b229f9 : SUCCESS in 5m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4652cabf52d4bed9834b510996afe47 : SUCCESS in 6m 02s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cacc96aeebf0458aad0eb65a29f359cf : FAILURE in 4m 50s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1bda453e17904e47a22c45af037fde04 : FAILURE in 1h 36m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/152d6b7d735e4879ac12ae2d59ffa527 : FAILURE in 1h 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d14369c65cfe968c8db8f1968915e80b7be4be7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-16 03:46:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1426613aa4c51af1e9aa5295b59f7c48e229fa66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 03:15:14.000000000","message":"Abandoned\n\nthe problem being handle with this patch: https://review.opendev.org/c/openstack/trove/+/886241","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84d574d7e1cee03240290315e43f8df125ff6719","revisions":{"f6baa6823a2905af623f37524b57323c149b0064":{"kind":"REWORK","_number":1,"created":"2023-04-06 12:31:06.000000000","uploader":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"ref":"refs/changes/38/879738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/38/879738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/879738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/879738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/879738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/38/879738/1"}}},"commit":{"parents":[{"commit":"76cffae9443c9a627f50ba5d54b187e5e8e90710","subject":"Merge \"Add missing releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cffae9443c9a627f50ba5d54b187e5e8e90710"}]}],"author":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-06 12:27:59.000000000","tz":180},"committer":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-06 12:31:05.000000000","tz":180},"subject":"Fix psql connection","message":"Fix psql connection\n\nFix creating database if DB is \u0027shutting down\u0027,\u0027starting up\u0027\nor \u0027cannot run inside a transaction block\u0027.\n\nStory: 2010561\nChange-Id: I4dcd2f4b556d3ea95b3562580d5a18899fb1100b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f6baa6823a2905af623f37524b57323c149b0064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f6baa6823a2905af623f37524b57323c149b0064"}]},"branch":"refs/heads/master"},"84d574d7e1cee03240290315e43f8df125ff6719":{"kind":"REWORK","_number":2,"created":"2023-04-06 12:44:15.000000000","uploader":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"ref":"refs/changes/38/879738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/38/879738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/879738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/879738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/879738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/38/879738/2"}}},"commit":{"parents":[{"commit":"76cffae9443c9a627f50ba5d54b187e5e8e90710","subject":"Merge \"Add missing releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cffae9443c9a627f50ba5d54b187e5e8e90710"}]}],"author":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-06 12:27:59.000000000","tz":180},"committer":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-06 12:44:02.000000000","tz":180},"subject":"Fix psql connection","message":"Fix psql connection\n\nFix creating database if DB is \u0027shutting down\u0027,\u0027starting up\u0027\nor \u0027cannot run inside a transaction block\u0027.\n\nStory: 2010561\nChange-Id: I4dcd2f4b556d3ea95b3562580d5a18899fb1100b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/84d574d7e1cee03240290315e43f8df125ff6719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/84d574d7e1cee03240290315e43f8df125ff6719"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
