)]}'
{"id":"openstack%2Ftrove~879896","triplet_id":"openstack%2Ftrove~master~I548eb43a3d5adbd84957b6b92d4b172bfdfb80a8","project":"openstack/trove","branch":"master","topic":"mysql-8","attention_set":{},"removed_from_attention_set":{"35169":{"account":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"last_update":"2023-07-04 04:56:11.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-07-04 04:56:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I548eb43a3d5adbd84957b6b92d4b172bfdfb80a8","subject":"mysql: explicitly use utf8mb3","status":"MERGED","created":"2023-04-07 14:28:06.000000000","updated":"2023-07-04 04:57:17.000000000","submitted":"2023-07-04 04:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879896-mysql-8","meta_rev_id":"533d395eb497a1fb6d502531dc65d21f4fabd930","_number":879896,"virtual_id_number":879896,"owner":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-04 04:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-30 07:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-04 01:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 14:34:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-07 16:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 03:36:24.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-06-30 07:26:08.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"a0002c15c4200a899e017654f8172129a6b191dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"date":"2023-04-07 14:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97b6c0ca9453e66b4c54da1eeac2365a10ddbcc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 14:34:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcde9aac92154ee28bde20e5805c05fb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8eb270f2857648668681109d58666e63 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24bc9cfcea964f4297c8b6da50df703b : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1a9d7c32932da011b9343442ef502466dc2ce9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 16:46:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2911ba40bb5434abfcfc7ce5f8c7a79\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/13848526a1ba49918ab8bc5ea8849a0a : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc0fc2cd01ba47b2a7b6e49c4514bb3d : FAILURE in 5m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37858b257025499580254b814d80a5d8 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d3ee338b7384377a8f795f45172db26 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a17cd5c02ba54373838685414e5928ba : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c941f50bb44a11a27bcaf6d960d98c : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e0748524f494b3cbaaec61a47b5d825 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95de0fe3138047a8801f1e92983fd132 : SUCCESS in 5m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dbc130effe924e40aee189d3e8b71f93 : FAILURE in 2h 16m 08s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c3611cfdad45404ba725f5698ab6fd58 : SUCCESS in 1h 49m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b35514a3a71468393871d3ed5f40452 : FAILURE in 1h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"499be4a4197a3e1b70d5a8343f46f87e8c71a929","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-16 03:36:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d29be8e2cbf63a4b9a04ca60065cdb29f20ec53e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 08:21:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e0a382ce2f0671d9c7b0b5268091f3ec8743d1b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 08:29:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a29e80e565334e5fab3883b65352a599\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35420fd56b12404babf8deb711c71104 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0b0e701b2ad4f61abee24d2fa77089c : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a75e336327a4831a73b8cf7382263cf : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1044ffabc8258b07bcaa574893b34400b5e89852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 10:29:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01119290c9d64355971b6a59cafe1b50\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b7baf7e9ecdc48cba4c506ee69dc984b : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65a81c3a25314cc9b37a764cfedbb78c : FAILURE in 7m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c451bfc0d4a64daf80862b394d715eaa : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec7e06d00d44ee59b59cd2b511ba2dd : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50d2b6a1e1a84e8d80d0f1dde1a1ece5 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c862919990942b295090a45747f932f : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a357e56c0d44e69c9cbc909bdf15f7 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19212f7863a342d6b4f924d8b94ad767 : SUCCESS in 5m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a1ddd2547ee4920a96a800519fc7024 : SUCCESS in 6m 28s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/fb440b9ff2364fdeb675f49245682efc : SUCCESS in 2h 05m 55s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5fe0d000b93f49b4a08637cda723480f : FAILURE in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f21ba47a57761ecc1653cdc42d8ee7f628e3390","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 07:26:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ac6d18035aabeca0afdaec5baac610afa166227","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-04 01:51:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1342e4d1afb1c84ce0c36504bde54f9c1adb1eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 01:51:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2bf356d710b306e57df3f8d3feabd6671cf3aee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 04:56:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d92d011c6f4c9da67beede31606283\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/903300d2a2a545218b1675a15dffc9c3 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9dc43f80364a98be13dcc033706833 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/144f80a55f5e4744ab4267c292643509 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e18520ef92548bf838ba84d143b2685 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9660abc8292d41b29e9699766e3294ea : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2cc5405b724274a02c555b0172838e : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25d69adfb2c74ec1bacc35c99a582080 : SUCCESS in 4m 18s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/56b9465540e4455196e8e129b2f2f5d4 : TIMED_OUT in 3h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"edaaebedc7fc443cf872f1d1589373fb84f42043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 04:56:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"533d395eb497a1fb6d502531dc65d21f4fabd930","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 04:57:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beb312869f92472cb966879d051c47fb\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d75ab1ce66854ec6996e6f08e308b14b : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3a4739342164883883d559fb652bf25 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d662627a2b594308a978943603460554 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ba1f0d955a446d96784e87411b38993e9ed7402","revisions":{"a8c629bd20d47dad71334e67e7001f3515997344":{"kind":"REWORK","_number":1,"created":"2023-04-07 14:28:06.000000000","uploader":{"_account_id":35169,"name":"Alexander epaneshnikov","email":"aepaneshnikov@sardinasystems.com","username":"alex19EP"},"ref":"refs/changes/96/879896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/879896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/879896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/879896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/879896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/879896/1"}}},"commit":{"parents":[{"commit":"76cffae9443c9a627f50ba5d54b187e5e8e90710","subject":"Merge \"Add missing releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cffae9443c9a627f50ba5d54b187e5e8e90710"}]}],"author":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-07 14:06:17.000000000","tz":180},"committer":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-07 14:27:43.000000000","tz":180},"subject":"mysql: explicitly use utf8mb3","message":"mysql: explicitly use utf8mb3\n\nthis fixes mysql v8.x.\nin mysql 8.0+ utf8mb3 is deprecated and utf8 alias doesn\u0027t work.\n\nChange-Id: I548eb43a3d5adbd84957b6b92d4b172bfdfb80a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8c629bd20d47dad71334e67e7001f3515997344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8c629bd20d47dad71334e67e7001f3515997344"}]},"branch":"refs/heads/master"},"3ba1f0d955a446d96784e87411b38993e9ed7402":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-29 08:21:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/96/879896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/879896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/879896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/879896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/879896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/879896/2"}}},"commit":{"parents":[{"commit":"7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8"}]}],"author":{"name":"Alexander Epaneshnikov","email":"aepaneshnikov@sardinasystems.com","date":"2023-04-07 14:06:17.000000000","tz":180},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 08:21:52.000000000","tz":0},"subject":"mysql: explicitly use utf8mb3","message":"mysql: explicitly use utf8mb3\n\nthis fixes mysql v8.x.\nin mysql 8.0+ utf8mb3 is deprecated and utf8 alias doesn\u0027t work.\n\nChange-Id: I548eb43a3d5adbd84957b6b92d4b172bfdfb80a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ba1f0d955a446d96784e87411b38993e9ed7402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ba1f0d955a446d96784e87411b38993e9ed7402"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
